)]}'
{"id":"openstack%2Fswift~767077","triplet_id":"openstack%2Fswift~master~Id1f337f400e182e0e1667d848a0cde744ec2c33e","project":"openstack/swift","branch":"master","topic":"c751966_md5_fact1","hashtags":[],"change_id":"Id1f337f400e182e0e1667d848a0cde744ec2c33e","subject":"Update for FIPSified md5","status":"ABANDONED","created":"2020-12-15 04:09:44.000000000","updated":"2020-12-15 15:41:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"511e41bad65b49c951628c26c8c9087956500b4c","_number":767077,"virtual_id_number":767077,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-15 05:12:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 04:10:34.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2020-12-15 05:12:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40814669126a575398d6957bda7aa3dd74e57e35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-12-15 04:09:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80d682ddcf5c871f65d1b8031a164f444a9d1275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 05:12:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e44032f864147a19727851c8f022abf : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/0edcb02c47d94823aa1a53e69d9b35f2 : SUCCESS in 1h 00m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/651be9887c9f4e0c9cc216c21b77b955 : SUCCESS in 45m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb354781d95c40b3ac7fea21aead1c60 : SUCCESS in 7m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d62c538773fa4d6cae09fae380d192b9 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12d6602ba6714749b97b46ac8385e351 : SUCCESS in 10m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e17083ec71ba472a9e042cde890bec88 : SUCCESS in 12m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c11db9e1e00743b6a3c8b6180e76162f : SUCCESS in 11m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c817476853554f6b9eb464dba551b98b : SUCCESS in 12m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/493e4570a7884f8aa54f6c183d4a3c8b : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe19aa91334849c8ada81528149e9ce6 : SUCCESS in 13m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce3201c118e04b6ba81ce8f6e445e753 : SUCCESS in 16m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/32b0ea28feb14caba051f42360ec6de6 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/01ef887034cb4eb7a115b98824f57609 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2fe103e35530488b8ddcb11e14f70288 : SUCCESS in 16m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15bca1a225df424d801455dfda7964fc : POST_FAILURE in 5m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/74a043d329f847678277e71e21931d68 : SUCCESS in 47m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad36db5e49df4e61bcb60f1ed95554e6 : SUCCESS in 40m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36716fa98e4f48b39a0cd776e2ee1dc0 : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbe8ff16f37c43fe80b29845109f41c8 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7081f3e4ae14ee59b56baf80bac80dc : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c084977aa194713ac2d9e143b4ae493 : FAILURE in 7m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a4b5ddf21cb4d0c97637f34a154e865 : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"77c32d087dc967f3289f20085e2b50909553afa8","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-12-15 14:56:26.000000000","message":"Patch Set 1:\n\nThanks.  Squashed into the main patch.","accounts_in_message":[],"_revision_number":1},{"id":"511e41bad65b49c951628c26c8c9087956500b4c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-12-15 15:41:10.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0de36c6dfbdb4eb200889c79290f575db91eef4","revisions":{"d0de36c6dfbdb4eb200889c79290f575db91eef4":{"kind":"REWORK","_number":1,"created":"2020-12-15 04:09:44.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/77/767077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/767077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/767077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/767077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/767077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/767077/1"}}},"commit":{"parents":[{"commit":"191a686a62683b6d5e35bb3c130b66a7a350b116","subject":"replace md5 with swift utils version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/191a686a62683b6d5e35bb3c130b66a7a350b116"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-12-15 04:08:35.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2020-12-15 04:09:33.000000000","tz":-360},"subject":"Update for FIPSified md5","message":"Update for FIPSified md5\n\nChange-Id: Id1f337f400e182e0e1667d848a0cde744ec2c33e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0de36c6dfbdb4eb200889c79290f575db91eef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0de36c6dfbdb4eb200889c79290f575db91eef4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
