)]}'
{"id":"openstack%2Fswift~767227","triplet_id":"openstack%2Fswift~master~I88c7a9fb78890302bb71aeaf63ad587e59767726","project":"openstack/swift","branch":"master","topic":"replace_md5_with_security_annotation","hashtags":[],"change_id":"I88c7a9fb78890302bb71aeaf63ad587e59767726","subject":"Remove md5_factory function","status":"MERGED","created":"2020-12-15 22:41:52.000000000","updated":"2020-12-18 11:09:57.000000000","submitted":"2020-12-18 11:06:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"767227-replace_md5_with_security_annotation","meta_rev_id":"8231bd9272d6e67aed1c7d51863e2da85529280b","_number":767227,"virtual_id_number":767227,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 11:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-12-15 22:51:35.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-16 00:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 22:51:35.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-12-16 17:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb11ed057c6d6f8a1e5835c55a50cfa9314cb777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-15 22:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91d323a17707285438b496cc061b09c5085c7dac","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-12-15 22:51:35.000000000","message":"Patch Set 1: Code-Review+2\n\ntotally","accounts_in_message":[],"_revision_number":1},{"id":"197a776b42805e8b44f382ffee7806b300c1d972","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-16 00:21:57.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a94b2a3cb1aefaf23ee6f1ac744188e9946f6fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 17:38:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b53c737fff72420c95f309c5cb77b09b : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5ff16abc6ac5453ea034a572ff49758e : SUCCESS in 1h 04m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65eac7094c8e47959b9e615c136a053b : SUCCESS in 48m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7511e2f7d32f437da467840d029ba549 : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/73cc78a5b77b4c619708bd2195418d80 : SUCCESS in 11m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8f98aab979f4a8987bed700d932cee8 : SUCCESS in 14m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3767be0b1d46452489f5d46c58c5ace5 : SUCCESS in 12m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/74c9979a251348debad598ca429a81a7 : SUCCESS in 12m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7b902b1f33e1438f8825577c3a03d135 : SUCCESS in 15m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31716cba71ca427e9995a058eea37af8 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/325cdc568b364351851aeabe39fa406b : SUCCESS in 15m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73766175433940be9555e26e8cfcfb3e : SUCCESS in 20m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/09dcbe83d2c4468aa4d7f1a8fe969921 : SUCCESS in 19m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4c9ff692fd06478bada6b875ba667c57 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/054f1a9c4dd04f01a9b63e7d8e7245b4 : SUCCESS in 18m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6020486fc524d54a0f7159f210e2cfa : POST_FAILURE in 4m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/617c611ed1c145509fab8ce79c67402d : FAILURE in 49m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5228a0546dd9451f8f8555233ccda248 : SUCCESS in 48m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e15fc693cb14c5b82453bf18bc1c9d4 : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d5a6256a62e46699bdcf0a9664079b0 : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b7a66e9b1cb4598969ace207f64aa00 : SUCCESS in 7m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/43984c64097842f191a4c8311a28aedc : FAILURE in 11m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fba9698df09e4edc93d282d882c1c570 : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a8b4dff10542b132cbda0a3dad477145c3452657","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-18 00:30:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49a1b8bd6706ef0cb5bc93452f30350758357fd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 01:46:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a563edf5f544689aa2bcba8fefb00f : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/72918883d87c4d769255215d91c853a2 : SUCCESS in 1h 11m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bee40b24ab2e49b584330ff4072ef903 : SUCCESS in 44m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f57dbd3fe6bd499585d6c5a119fdce9a : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c758362a5e0841f8add226c9add7f9b7 : SUCCESS in 9m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/783cb09f7c60456e8d675d1ef7dfc8cf : SUCCESS in 12m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e4baf8e916a84d70b17923838bb91c17 : SUCCESS in 12m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bbb74289c05414aa47857bfd359ce95 : SUCCESS in 13m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/096e4882d8fc4184bec2b9321d5688e1 : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/459172df4f934c379650f213783bc62c : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdd11991d4ae4c6587f4d1d935e4db78 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98a39b859ec4403e8362442ab676682b : SUCCESS in 18m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b2ccba08542c412d8d310c94bb883b28 : SUCCESS in 24m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/06b85be729794d0ba4f8d296017fb200 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/50d22b7a10024edeb2a721e49a80182d : SUCCESS in 18m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d8705b3178e4144943e2df4879708b9 : POST_FAILURE in 4m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2969dafbff614e4dbb976935f7ef6c15 : SUCCESS in 44m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b224f8802ee442c09218ed8fe24a7633 : SUCCESS in 39m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/121c489be8804ea5b1c84fadae6c3d90 : SUCCESS in 38m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cb738f987304af7a9e30a3bb64e6455 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8fb01eff05478c8f078b221de4059e : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ab013f576bc4713bdb9db94a0024d72 : FAILURE in 8m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50a8dcfd1bc2456f89bb9e2095472bcf : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5dba4061f43cb5316aea81145eb4fcb851b24957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 01:47:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ea7b7397fe44e66dce2c07cf5d8d057a669e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 05:52:53.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"2d3e9ff37ffc65a062d7ce3cd53b068bce6b1b1d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-18 06:54:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2fb48cc77f29f09c4c15e2e47cd310323df0754b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 09:11:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e23438b51e45fcbd544e284ad9e0e2 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a55336e9d9504178aa6b71587255e97c : SUCCESS in 1h 11m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ff9e78d6fa84f0e83822dfa9c559642 : SUCCESS in 41m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0db817aa3a84e5f9e223420181cca88 : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cd688a30aaa84a58a038d0509e7b8d62 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f09fdd596ea4d3f9233cce98de6e02b : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e39404d7a78f47c7b3751e3eb35bd549 : SUCCESS in 12m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/27707f10a7e74cf5ac863aa4c8af7d6c : SUCCESS in 12m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5ff369c128af4e9f8b35bb1adc92975f : SUCCESS in 12m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b15ff57980a495aab040e6015b98bc5 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eea4c35c8ea843d080c9bbfbebfd765d : SUCCESS in 15m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/739c5b66a39c42e8ac96157e1e58c7a7 : SUCCESS in 15m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d95b566868f24b0f94fbf4585ea313f4 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8701d80bb92d4191a20255f20dfd8dad : SUCCESS in 16m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/67cc53d2e6d44d41b58de9149272ba8d : SUCCESS in 18m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db80f2750ccc4d049c45f87d8c424a79 : POST_FAILURE in 5m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/49c982939c3046ce9327ac2cb7be6b78 : SUCCESS in 44m 40s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/58b6eef7deec40b681f0979c65e73807 : SUCCESS in 45m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e523bc590e242b59d535665fd44e4ff : SUCCESS in 34m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74be1ab1ef2e4cacbf5b529f5e1b158a : SUCCESS in 36m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2df01a276a5447078473f4bd543e2205 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e0808b9a5940e28363b2e72880e73e : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff7a727309d3424cb094fd2ff25180c8 : FAILURE in 6m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38a37acccf87498b9696d545b62eec69 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f1216f4a0859a896ee00c8fbc48d05315cdd3b78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 09:11:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1f026349b666a40250f3b6f16231ef31bbed6ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 11:06:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63407d0b43b64849b42a1e7d209017f9 : SUCCESS in 17m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/98ab174da05e4679a1434d2bd7317e4b : SUCCESS in 59m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29b87e726f2d41b6a5028bc28b237c26 : SUCCESS in 42m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0c05f35e2f854b78906aeae05384f04a : SUCCESS in 8m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/963edb9eb70e429d90adca86ce89e3f6 : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b1b7f3e5e0345a8890563d24629b193 : SUCCESS in 15m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b0ae758df0c4ff091962a2eb83b1fc3 : SUCCESS in 15m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4e750711aa974e12b421314400526da6 : SUCCESS in 17m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b28dbb4796f40e7971e469b98ea2077 : SUCCESS in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81ef948b8605492b939d7c82647983f7 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6e02fe14a293418695cc27ba09f3527f : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89d49e01a0f5431c945ccb61a68a681d : SUCCESS in 18m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/30f43fe537a445bc84d971227a358aef : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/665d59674d354cd18b2bd33c12a6b97b : SUCCESS in 19m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/976b52952ea242c6a841d0bb7503b2f6 : SUCCESS in 22m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d4c157914f14464ea4f3165b3ad72369 : SUCCESS in 1h 30m 52s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/827af62116bf4695bcedf9339af1bef8 : SUCCESS in 41m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4e1ed8e25294bf49e26da7639176304 : SUCCESS in 43m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ab7534aecfe41eba4de632dce487ab3 : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b6bcbf5274d4f44b0fb4b0dfde25454 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb11cdd497b487dae4eefca3f7299dc : SUCCESS in 5m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4f05430e3454f15afd377a83975ed69 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b1dd3478221ccb0b7e8f492b6430671d90221f6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 11:06:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8231bd9272d6e67aed1c7d51863e2da85529280b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 11:09:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37ae3f388764269b8075fa7b7c7fb0f : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1ba5d2c6714b4180be1f83950151edcb : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/45ad6d6e044c4833a3dd746c7bf23a39 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84adf5e3bc19f42b9687b6adcb8fb14bd43595f9","revisions":{"84adf5e3bc19f42b9687b6adcb8fb14bd43595f9":{"kind":"REWORK","_number":1,"created":"2020-12-15 22:41:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/767227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/767227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/767227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/767227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/767227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/767227/1"}}},"commit":{"parents":[{"commit":"5320ecbaf2c0e77842ab1ee3eb8106948dc06704","subject":"replace md5 with swift utils version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5320ecbaf2c0e77842ab1ee3eb8106948dc06704"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-15 22:39:44.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-15 22:39:48.000000000","tz":-480},"subject":"Remove md5_factory function","message":"Remove md5_factory function\n\n...because otherwise I\u0027ll find myself wanting to just use it\n*everywhere*, which seems to defeat the purpose.\n\nChange-Id: I88c7a9fb78890302bb71aeaf63ad587e59767726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84adf5e3bc19f42b9687b6adcb8fb14bd43595f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84adf5e3bc19f42b9687b6adcb8fb14bd43595f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
