)]}'
{"id":"openstack%2Fswift~767243","triplet_id":"openstack%2Fswift~master~I9eb84d1c794d58f0af3b7d78d3bc4660c1823dc8","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I9eb84d1c794d58f0af3b7d78d3bc4660c1823dc8","subject":"s3-tests: Only try to copy outputs when there *are* outputs","status":"MERGED","created":"2020-12-16 00:31:31.000000000","updated":"2020-12-18 10:41:33.000000000","submitted":"2020-12-18 10:39:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"767243","meta_rev_id":"b9fb1176cfb23be8e833fab7c3117841f44a0a0b","_number":767243,"virtual_id_number":767243,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 10:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-12-17 20:32:23.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-17 18:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-17 18:58:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 18:08:47.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-12-16 18:49:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c73faf601b02e5c80993ee45a79baec875a14d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-16 00:31:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44bca9ba93c0f07c05d476503a53584dc85779c4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-12-16 18:08:47.000000000","message":"Patch Set 1: Code-Review+2\n\nThis makes sense, and although I am not sure this is the best way, if you say it works, I\u0027m okay with it.","accounts_in_message":[],"_revision_number":1},{"id":"22bff046545010c4e5166d64d3ba9188ffeef984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:49:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f1952609724ceb951a6c0adcb5e920 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a1a9ccb9f902435cb25e80b3d6477938 : SUCCESS in 53m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89bd0a954f2c4055b6e9e42f01fc7d41 : SUCCESS in 56m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69df19c5bd0a49caaf160328022c184b : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c0040e99a514eb29c6235ede2b9e608 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19dacf16ed324d01ad3d41c22b1fcc76 : SUCCESS in 14m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09a0d5c0eb6f417ba8058075c0f67f79 : SUCCESS in 13m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/303b9aadaa994d7583afcf44757eed14 : SUCCESS in 13m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d745150494ca49d98536b78afec1ad5c : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9488b90ec1974dd181be23159ef5dae4 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ac51f627df7421c9300b58da52bb225 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47c0a041c657463ab9a2de935d34f225 : SUCCESS in 17m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/46cba6d4baea430f8a98d1036eeb140f : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0e966369dce74b7db8c6487d85300d92 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5c732f64f6164b598e385088df12cecb : SUCCESS in 20m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/638f01767a7a4c0294f35c111340a285 : FAILURE in 4m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/af52dcd182a34e79a6b13bc8e5abe8d5 : SUCCESS in 47m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c5f7dc7d5194ff9883602c729883386 : SUCCESS in 41m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5da079724ec42388b2ae922aa9ea14c : SUCCESS in 41m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94a732f596f34d2bb3401cd6d506d64b : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69a8e9659f348e580e998791760c50e : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9dafe0117cb14dcd8403d74fd51de6c2 : FAILURE in 8m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e582a8c0ff643f38f08c51b2c6919ac : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7bc9dd972328df68efb7591bfc774d570e1a7c0e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-17 17:12:23.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecc0786f86790bb1899124af4b1e5673999b59af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:12:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2a4d8036f600249838e5754e23bc8e53a5fec20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-17 18:58:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"644baa1741b72481a94bd1c54829ef5c6097d0b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-17 18:58:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nJust fixing up the commit message since the gate job was failing anyway.","accounts_in_message":[],"_revision_number":2},{"id":"9301dcd79d48d8f6c0c57ea2811704b3f413d810","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-12-17 20:32:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7d9b25c512ea2d57865d5578c39d594c4f958cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 00:34:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f76f50f76849fe961d21d00dbf67fb : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/cd70a0f9622040408178f3bcd42a5448 : SUCCESS in 57m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e8723949acea4c1aaa8b4a0da30efae7 : SUCCESS in 47m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a64b1340efee4abfbc49a1437572b7f6 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/215e2d55250c4890b35281fd8081e8ce : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4033b5f4735441d9976cd962a9f2ab1b : SUCCESS in 13m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/096060e3b510423696858a439960fb8b : SUCCESS in 12m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e905df3af634ec4ae67c19f35edd761 : SUCCESS in 13m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e07e7a27dc5f4be78b6c75b1c5cfe540 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a6cb14148db4b788c78b4737230632f : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1c176a7e21824d9a8296abb90262cc05 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ee84955256540ca94665a569b14508c : SUCCESS in 16m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3225c112d77645d88071e252a3352ee9 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d202b9ae75b24005962540b173d06e52 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6c4cbdf1677b414e9065c7cf2d5a6e2f : SUCCESS in 21m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df1a7302b07a488ba45f124e186f5c8c : FAILURE in 4m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/26490a1e2c2e440ead266f3e00c1a261 : SUCCESS in 1h 18m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca226c6a33aa4eca8b1738bc13558773 : SUCCESS in 41m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1a54d74ccb541d5a956b89f1c96a5ee : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39ec3be7ad804f409719a25353ebefd0 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8264797522dc4379a95c06f8b6a7d433 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dad531615f2c4fc9bafa9148b56591bc : FAILURE in 10m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/860cb03b4e2b429abfb64c86abbfc754 : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8c9c351c80e2137bd49aa612eb44f0e78bf97ec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 00:34:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f45d9e0c8ec013b9c3401c4c0421e3ab5838238b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:26:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a38ae23b074da0911944ff34d8612e : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/094deafa87c2418082e28c23c733e6ae : SUCCESS in 1h 01m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfba37e84e6e402382e6fbb3706ac9cd : SUCCESS in 40m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b53defa5ab3b4411aa32ad6924b7cbfa : SUCCESS in 9m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/23f095b3df3947d8b6b47a02f1173e87 : SUCCESS in 11m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/401927f361ed4f2cbef0d91e7031bd11 : SUCCESS in 11m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d179de17386459d9b244a8e0a92f66d : SUCCESS in 13m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bca79561b1824749b67d70e13bce8a30 : SUCCESS in 12m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b4215effc24d4e218f5be651ea6c9194 : SUCCESS in 13m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a82cc0caa6ea48369b65d9d9739319be : FAILURE in 15m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a67386cacd1643348faeb83ae1994de7 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17927723f25044da9250a1929c0a5d3c : SUCCESS in 17m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e256d538edb743be903b5203c58b7d88 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/313b8bcd380d47afbb3c9e95cfa6a122 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a805d13a6dea4a408fcde7062073cbda : SUCCESS in 16m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf12dc1148eb451caf915f31e5bde592 : SUCCESS in 42m 22s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7ee149af714a4186abd4a256c360e6a7 : RETRY_LIMIT in 2m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c60545694d94ebb97e99f135a22f6df : SUCCESS in 35m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82cbb7e5be494bb3ad4f790da8036452 : SUCCESS in 37m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24efd5463a0b4f4fa2ef1884e037be09 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f1d69d51484253bd7651aef8f6f4ab : SUCCESS in 7m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/844b5f29e47a4d688029e92420a93d89 : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ee4ed48e3d649b5c5a8a3c3726d512142c9b1292","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-18 06:53:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"010c774c043dbdffbe73f424a87a48b62b1f5b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 08:45:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dcafdb351854a098f601d26fe8b11fb : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/09191e08266c482b9b86789a548b98b2 : SUCCESS in 57m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4be2e99b9fd343c48ef391d077e3b79d : SUCCESS in 50m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cdcf8d797b034bf59ca446dabe65b007 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7dcdbe8edb88435da49cffc770aadb2a : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fae96293af63469c98c2f075ec049a02 : SUCCESS in 13m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c20b5dfa2f184373a61f355d9a17afed : SUCCESS in 12m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/42188aca82b94f61bc0bb3e754e9e3de : SUCCESS in 12m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2c8964d96f604710830dcf444fa26a5f : SUCCESS in 11m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4763f21ea4bf4994a8802bf9e66c7221 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ac3d24e88cf4f358f0780831ed716ff : SUCCESS in 15m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/72ac9c014ab345bfa1d2a38f93122d07 : SUCCESS in 18m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7ef355f31d034e5b8b1e45bd30d63b1d : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9e2a913c9ebd4822923b0f76a35732de : SUCCESS in 16m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ae0ae9660be54494b00f800d86da1d53 : SUCCESS in 18m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/922f0117a0f14b7d91dd50eb5825230f : FAILURE in 4m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/282e350d3c0647c9b0b8d665a78a6431 : SUCCESS in 47m 45s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ea8289237bba43b28067cec7267c17c8 : SUCCESS in 42m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da378a9b2aef4b2cb6180b4a81f503f3 : SUCCESS in 43m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6dd3befd69ae4c58b16bf32e47cac471 : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d22639c9184458eb055ea79e880af15 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfad994747e4a80910e5346990c8d4a : SUCCESS in 6m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ffcf1042e564158b0d61792e27a8583 : FAILURE in 8m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71a7f665413d45ada8ddca121e2936f1 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e80411195ccc52fd9667384c842fe43c44a3bc51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 08:45:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d0a89ec16abbb7aa258e9679b6ae04ff72a960c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:39:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af417c08448841a7af7b5256c7222990 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/234da9e9389740a09a855ab2a54ec484 : SUCCESS in 1h 00m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c45e554bf3b42189261cb68b2306850 : SUCCESS in 52m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1d671dbbaa5e4db29ef256e77ad614bb : SUCCESS in 10m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8b3503369c6f41298827a5f93c758f6a : SUCCESS in 11m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bc88c53e15042cdb76a914a96a407b1 : SUCCESS in 12m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a25d93beccc24c1f8a52b6c38a0835f5 : SUCCESS in 13m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/33db0aebc161401199fd3e11b7f633df : SUCCESS in 14m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3c02bbf458b14ac4922466165c07ec13 : SUCCESS in 13m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a7b149aac1348a6a17435d0be25f1c5 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d212c9e7dd104a1ab492d2d81105cb51 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c260bd7764d4a098ededc7f500fcf6d : SUCCESS in 17m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bf8321ac3b4c4e60a1379799b5b6ac57 : SUCCESS in 27m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f3f8e0ba2b8d4b409e83e4e59d79c1c9 : SUCCESS in 17m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e9cd1dd22d6347c0b18c9711bef21084 : SUCCESS in 18m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e83483071894af89dc48cd7305e01fb : SUCCESS in 49m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bd66e14c245a4b5b921aef36c7994616 : SUCCESS in 48m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4cccc06c930240438231ac8cafb8530e : SUCCESS in 41m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da8b25927d1743738699cbcd010331e1 : SUCCESS in 41m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/128d46dc34fd4c11872961db0b57e11a : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b60fda7da1649708e41c2a3ca2603f4 : SUCCESS in 6m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fa4e1f5735b4b60be6e4d4126e4562a : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":2},{"id":"30e33e0bf55920d1d00bc3a5e2e6ef5c903ce56b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:39:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b9fb1176cfb23be8e833fab7c3117841f44a0a0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:41:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61e9c31655be47b581bb6bc33f79069b : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c2a8ccb28e9a41efb743b359677112e9 : SUCCESS in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bf5af66d267841df8a60a4ccaacfbcae : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73aa48a823a2e602089c3e37d6b9ec4c9e19d35f","revisions":{"40159af0b572a57203f48b51329c459e7daf8e99":{"kind":"REWORK","_number":1,"created":"2020-12-16 00:31:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/767243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/767243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/767243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/767243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/767243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/767243/1"}}},"commit":{"parents":[{"commit":"28cef6767fe28062563d1bb18722184139df50e8","subject":"Merge \"Fail if swift binary are missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28cef6767fe28062563d1bb18722184139df50e8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-16 00:31:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-16 00:31:15.000000000","tz":-480},"subject":"s3-tests: Only try to copy outputs when there *are* outputs","message":"s3-tests: Only try to copy outputs when there *are* outputs\n\nThat way we avoid POST_FAILUREs when the real problem was in pre-run.\n\nChange-Id: I9eb84d1c794d58f0af3b7d78d3bc4660c1823dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40159af0b572a57203f48b51329c459e7daf8e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40159af0b572a57203f48b51329c459e7daf8e99"}]},"branch":"refs/heads/master"},"73aa48a823a2e602089c3e37d6b9ec4c9e19d35f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-17 18:58:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/767243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/767243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/767243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/767243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/767243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/767243/2"}}},"commit":{"parents":[{"commit":"28cef6767fe28062563d1bb18722184139df50e8","subject":"Merge \"Fail if swift binary are missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28cef6767fe28062563d1bb18722184139df50e8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-16 00:31:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-12-17 18:58:09.000000000","tz":0},"subject":"s3-tests: Only try to copy outputs when there *are* outputs","message":"s3-tests: Only try to copy outputs when there *are* outputs\n\nThat way we avoid POST_FAILUREs when the real problem was in run.\n\nChange-Id: I9eb84d1c794d58f0af3b7d78d3bc4660c1823dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73aa48a823a2e602089c3e37d6b9ec4c9e19d35f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73aa48a823a2e602089c3e37d6b9ec4c9e19d35f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
