)]}'
{"id":"openstack%2Fswift~767319","triplet_id":"openstack%2Fswift~stable%2Fussuri~I09143ae20bd6249083c0b80cdaa9d561e6abb301","project":"openstack/swift","branch":"stable/ussuri","topic":"bug/1079075-stable/ussuri","hashtags":[],"change_id":"I09143ae20bd6249083c0b80cdaa9d561e6abb301","subject":"Fail if swift binary are missing","status":"ABANDONED","created":"2020-12-16 10:57:20.000000000","updated":"2021-01-19 12:46:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":764137,"cherry_pick_of_patch_set":6,"meta_rev_id":"d6ba795081b4a6694177d74dc51599bf70a6a7a9","_number":767319,"virtual_id_number":767319,"owner":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-16 22:38:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":-1,"date":"2021-01-07 21:27:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 22:38:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 21:27:53.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"1d7389fea9053f05196917a461c1eaf02264bc19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-12-16 10:57:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0ee067d8d2c1140d8c9f3d3f3fc84b7a0d9cdee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:38:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a589fddeedeb446b8fdd0f80673c66be : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/475affca0ef6401fbbefb5b0fdc73d21 : SUCCESS in 1h 17m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0651d4c4e85e4348b457dec4324b1d94 : SUCCESS in 45m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b6503d5471f4e3f8b45cc17cac40e32 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa53942df015488ab4c712b53acfa42b : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dd4b1134e92b45778a7ad68becfabd26 : FAILURE in 3m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad799cf1aac6443287540d56c252fe76 : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/19dd900548c340e1888e47ebdabd776f : SUCCESS in 15m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/abc0f29efb9b4786bae141d0943247fc : SUCCESS in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3cdc6fac47e4a68ad028a0826089edb : FAILURE in 3m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/566a9b9d39b44e8a8b0ab1a78917955c : FAILURE in 3m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3ff0c60ed9e648909ba68aea7148fc7e : FAILURE in 3m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b15d4b9dc82540e080741ad9a011795b : FAILURE in 3m 09s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/305ca83ee7364abb891489ec06dd6166 : FAILURE in 3m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/25c9b8d1c7c34e01b09f1a1e43d53657 : SUCCESS in 11m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7f070e1d310940c3b68cea53c98c7a94 : FAILURE in 13m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/56563d08e1e74deb909305988ae3944a : SUCCESS in 10m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bf9dbcaa8431414390212be4665b7780 : SUCCESS in 12m 12s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/93d3671981344bb3ba85e404473d6baf : SUCCESS in 14m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c381ca423c34910921f0210934cf2a4 : POST_FAILURE in 4m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/99eac8ac9929414cb09fea1c070b1f16 : FAILURE in 41m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44d1597a68c04a93b9e2b1b261383936 : FAILURE in 16m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/895d6d1d5da541d48e1a4f36f6e08fae : SUCCESS in 30m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22cb881822b74fc0ad55cc283ed7b77a : FAILURE in 22m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/581c9e134fc941ceabb7980141b4924f : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d25bcd73e6c4a1dbf6e8826003fc41d : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11b2a576658f431aacf8744429e90f30 : FAILURE in 8m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a5656e766c8437e9149a5853e580bdc : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"81964bda9a0aecaebe35288025d76d8724c3e65e","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-12-18 00:07:25.000000000","message":"Patch Set 1:\n\nIIUC https://review.opendev.org/c/openstack/swift/+/767014 needs to be cherry picked before this patch can pass any test","accounts_in_message":[],"_revision_number":1},{"id":"94bca2987b463e673f98657c92ee3d867044722e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-01-07 21:27:53.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"d6ba795081b4a6694177d74dc51599bf70a6a7a9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2021-01-19 12:46:06.000000000","message":"Abandoned\n\nCauses regression,see ttps://review.opendev.org/c/openstack/swift/+/769848","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b5dd37bba4a5e45881163421e56b3e1d87ff16d","revisions":{"3b5dd37bba4a5e45881163421e56b3e1d87ff16d":{"kind":"REWORK","_number":1,"created":"2020-12-16 10:57:20.000000000","uploader":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"ref":"refs/changes/19/767319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/767319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/767319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/767319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/767319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/767319/1"}}},"commit":{"parents":[{"commit":"a465c464cb227682f0054258790b3e4978293204","subject":"Merge \"Authors/ChangeLog for 2.25.1\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a465c464cb227682f0054258790b3e4978293204"}]}],"author":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-11-25 08:44:43.000000000","tz":540},"committer":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-12-16 10:57:20.000000000","tz":0},"subject":"Fail if swift binary are missing","message":"Fail if swift binary are missing\n\nCheck if the binary file exists before attempting operations on it.\nCloses-Bug: #1079075\n\nChange-Id: I09143ae20bd6249083c0b80cdaa9d561e6abb301\n(cherry picked from commit f4d691da8387af3e4402c20b92149cc4290cb2e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b5dd37bba4a5e45881163421e56b3e1d87ff16d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b5dd37bba4a5e45881163421e56b3e1d87ff16d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
