)]}'
{"id":"openstack%2Fswift~767320","triplet_id":"openstack%2Fswift~stable%2Ftrain~I09143ae20bd6249083c0b80cdaa9d561e6abb301","project":"openstack/swift","branch":"stable/train","topic":"bug/1079075-stable/train","hashtags":[],"change_id":"I09143ae20bd6249083c0b80cdaa9d561e6abb301","subject":"Fail if swift binary are missing","status":"ABANDONED","created":"2020-12-16 10:57:40.000000000","updated":"2021-01-19 12:45:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":764137,"cherry_pick_of_patch_set":6,"meta_rev_id":"9cf2f9961bce28a91859b0e940a75ff96f638efe","_number":767320,"virtual_id_number":767320,"owner":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-16 22:30:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":-1,"date":"2021-01-07 21:28:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 22:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-22 20:37:57.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-01-07 21:28:05.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"a6b34ae32203ff06bba0d3d492bdf2bbda6f1929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-12-16 10:57:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4b6fffe9fb854934fdeb85b5b4774879efdc094f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:30:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9412f33456984633a0425effe1664e95 : SUCCESS in 4m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/69c83ec3722348838350cf87e3449d24 : FAILURE in 47m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d30b560d2ea7438dab81d484d1addaec : SUCCESS in 51m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a69c69acebf74f96a60c4f9f5465c8b0 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/74e98de649fe4869bcd574a51a84ea1c : FAILURE in 3m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6d2885b0f3947778e94b23714a4499d : SUCCESS in 11m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0d36c665b8d45719f6a4953e375318b : SUCCESS in 11m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f2dab5e71c74cb49f578f2ca64de0ef : FAILURE in 3m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8aadc805c156455396e25a74b444f6a8 : FAILURE in 4m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a7b5243115464ba09a9dd1a4ef6701ba : FAILURE in 3m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6103c03c2d884d419fb7ab486dff6690 : FAILURE in 3m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6fa2eb73ae284b5da7acb41b6010bce5 : FAILURE in 3m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d5d9f6a2535342edb80534bb28cc8866 : SUCCESS in 9m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/89fbc88008c04f2e93d170112517ea7a : SUCCESS in 10m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9a005ded6abd4e31ad271c49a56189db : SUCCESS in 9m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/09403437d86c488aae9ab1e4803fa0e6 : SUCCESS in 10m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9f781b2a4c704f6b92f9361e33ff5f60 : SUCCESS in 14m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fde606329e634778a7d93e7d803a5865 : POST_FAILURE in 6m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c24931a0839a41c3836018de8ea5d831 : FAILURE in 32m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46a2f3ffaf2c40e2ba5fdfc51698cda1 : RETRY_LIMIT in 9m 11s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f4d281f676bc4cc0a303790f564fc0bb : SUCCESS in 30m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a44bfcbf30148c98b626f88b3254c76 : RETRY_LIMIT in 10m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d5c51f03bf948f590ea979937dda0b1 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19eba9e05754b8eb72e08e76ccc9430 : FAILURE in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aba2581638ce4ef7b906ff2823fb8559 : FAILURE in 8m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/226f4990c4e448b08b0eab5381a0bde5 : SUCCESS in 57m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/84c212a2fbb34ef286a03a219080e174 : FAILURE in 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"68ff93ef1fc2958c46929e44b6262f08f09384e4","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2020-12-18 00:18:44.000000000","message":"Patch Set 1:\n\nThis patch can not land until https://review.opendev.org/c/openstack/swift/+/765883 is applied to stable/train branch","accounts_in_message":[],"_revision_number":1},{"id":"eb6aa8faae1dba7cf7950a00d8e26cf97bfeffff","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-22 20:37:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9327cf28e5fb8cc5d807d408bae65a300de18311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 21:39:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffb331673de4a5dbc25e7af422dd7f8 : SUCCESS in 5m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/05a3225be93d4dfb978c7f84d0ab90bf : SUCCESS in 59m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c47183124b6451e9c59587c4520a980 : SUCCESS in 37m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4dfb47d1412848c09ed076b88f43015c : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4daa6aca612b4d8ebba00d4dc8ecc159 : FAILURE in 5m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ba3231f9c0a4c41983dc4a942140158 : SUCCESS in 12m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/46f03d2f2cba4fe0b9ea48aabd9cf88d : SUCCESS in 11m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af639f3a0d1f4bce84f56c516f276cf3 : FAILURE in 4m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/80af70b101b448f683d9eda2a3e9acd2 : FAILURE in 4m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9f5142d1e52f4ba7994d94b6c38690ed : FAILURE in 3m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6254f15dedd54875870bf0c954e36718 : FAILURE in 4m 49s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5c9635c6331d4979915252fd78930f64 : FAILURE in 4m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6ad36f7bd62945fb9e4cfcaca78e2950 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a9e6647bd3074783b05281464e9fbc96 : SUCCESS in 15m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4bdf93ab022948e39ced10ec9402c27c : SUCCESS in 10m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/babfacd930f64cd49a0417c6f2640d18 : FAILURE in 11m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/54d044352813448dbb20fa72cd18a3f2 : SUCCESS in 14m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/843575ca4a4c42b8ad79ac929f9c246c : POST_FAILURE in 5m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/67f60de6395a4309bbecdb7bae4927d6 : SUCCESS in 32m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a82e8941616f42c8ab81f9c21b3a283c : RETRY_LIMIT in 16m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/eeff50f40cd74876a64158c2372b57d9 : SUCCESS in 32m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1d7b2d779334c0586aa9cea7d23463f : RETRY_LIMIT in 13m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a0ff05c6896405b9cfb1bee15a52400 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd319b723d1f4bdf95066b83ab885e90 : FAILURE in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3deb0eab2a14e37bd5f26dd800af7c2 : FAILURE in 10m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd9819c18b3e468e963f28aea2f6d7b2 : SUCCESS in 49m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cf4a289a0d964deca303a217150bad38 : FAILURE in 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"38f1c158c6ae8e642eeb05430b56799a7de38676","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-01-07 21:28:05.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"9cf2f9961bce28a91859b0e940a75ff96f638efe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2021-01-19 12:45:52.000000000","message":"Abandoned\n\nCauses regression,see ttps://review.opendev.org/c/openstack/swift/+/769848","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e773965d8eb44b4b67c4177156e27496712fc42","revisions":{"4e773965d8eb44b4b67c4177156e27496712fc42":{"kind":"REWORK","_number":1,"created":"2020-12-16 10:57:40.000000000","uploader":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"ref":"refs/changes/20/767320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/767320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/767320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/767320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/767320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/767320/1"}}},"commit":{"parents":[{"commit":"ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1","subject":"ChangeLog for 2.23.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab921f22f628b40cd6efcc9dc77ccdaf9b3ec8f1"}]}],"author":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-11-25 08:44:43.000000000","tz":540},"committer":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2020-12-16 10:57:40.000000000","tz":0},"subject":"Fail if swift binary are missing","message":"Fail if swift binary are missing\n\nCheck if the binary file exists before attempting operations on it.\nCloses-Bug: #1079075\n\nChange-Id: I09143ae20bd6249083c0b80cdaa9d561e6abb301\n(cherry picked from commit f4d691da8387af3e4402c20b92149cc4290cb2e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e773965d8eb44b4b67c4177156e27496712fc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e773965d8eb44b4b67c4177156e27496712fc42"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
