)]}'
{"id":"openstack%2Fswift~767644","triplet_id":"openstack%2Fswift~master~I3d6e2e4542a5ed03a6d31ec0ef4837d1de30a045","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3d6e2e4542a5ed03a6d31ec0ef4837d1de30a045","subject":"s3api: Better-handle SHA mismatches during CompleteMultipartUpload","status":"MERGED","created":"2020-12-17 23:19:15.000000000","updated":"2021-01-08 04:53:58.000000000","submitted":"2021-01-08 04:52:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"767644","meta_rev_id":"f330e6a360ed2fd691210d2b0e04cc23b410593a","_number":767644,"virtual_id_number":767644,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 04:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-07 19:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-07 19:15:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 01:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 14:45:03.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"e8daa3fc800ca33a530027cec3290b48eb6d7427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-17 23:19:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f021bdd6a19baa36c602c315bf39002d866a1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 01:01:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e401703770432fafc8753228b23fec : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/aed8b9583ee84c1c9a00baeedb10c2e7 : SUCCESS in 57m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b12d1234a979436189d599627efe8bc7 : SUCCESS in 47m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5492897346ab462082d075242e9e5f0e : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e462bc311984c699b582ec55c693e40 : SUCCESS in 12m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b75130a7dc8443f4993e96a6fcac9348 : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b0a79061da3464f9dc7d1acbb406e74 : SUCCESS in 13m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a1a1519ceae64dd0b78c82586ba1b50e : SUCCESS in 12m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/74ea0ae540704879b4924c417e2c7525 : SUCCESS in 16m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b2963bc7ddcf4114ba72d3cfb7abdcee : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa16694fa984488c8920160901ef0160 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d69cd72b79ad456d8b176b31a9b19898 : SUCCESS in 16m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8bc4d159f83e4aed8e69029d99a7d104 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/057ec40cd4794558b3b9d4a2a3311ca9 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0676e3b54ec040f2a1006e0db108366d : SUCCESS in 23m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/05c236302bc14d19b439cb6e63540975 : POST_FAILURE in 6m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ee1f1c5ad4e6490c996b70cbfa12c386 : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b96190bc7f04bc9b8cb37db2080924d : SUCCESS in 37m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/032674a2ff1a4ede87fa69d0e152faa0 : SUCCESS in 37m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7015e88be4834b6081f7b652b7d585ab : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61aa9991cdfb44988e480cf7cab7a49d : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e1811040aaef44c3bbad74c2644cf1a2 : FAILURE in 9m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/141c642502ba4e7f821b5e32b03397d8 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b7ec4aa8b9df4e5cba772d9b74132718f2669c66","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-07 14:45:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nLGTM but wondering why the fix cannot be made where the original exception is raised","accounts_in_message":[],"_revision_number":1},{"id":"a09e893cfa8ed705663f07a8f143560c35cc9ecf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-07 17:31:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c28bc1fc3ab4646369772e0e5425dd17f0e8c0b6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-07 17:47:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nok makes sense","accounts_in_message":[],"_revision_number":1},{"id":"9546eda66c0a087760fc55dc76472628b8f42bf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 17:48:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d38082befabd14ae84c91ce30d937f2d7b28b369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-07 19:14:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfd07ee21cb0e8314e125ffbf0a7d6e086609a83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-07 19:15:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nNeeded a rebase and an extra import following https://review.opendev.org/c/openstack/swift/+/751966 -- approving based on prior approval.","accounts_in_message":[],"_revision_number":2},{"id":"a9cc1ef7ccffb077302a7bf500b2189675382250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 23:53:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff2c08452a94701aa6a0f6b76dd213f : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/871a67bccc094a5586efd8df9a274d7b : SUCCESS in 1h 10m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eed59feea5f94f2a8f55ee4941ab55c1 : SUCCESS in 43m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0832698b2a14c54b05d90e88b1ca648 : SUCCESS in 8m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/42485ffe3bcf4ba0b37d4e5ff3f27e34 : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7b9232debbf4abb9a6ffba45d7e97b8 : SUCCESS in 13m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6cfb4dd3640547ad8c1ce81c1b4ddba4 : TIMED_OUT in 1h 31m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a1be988ad44946aa94025f7854988557 : SUCCESS in 14m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/386a3c830d704bb7b1a046ac57f2bba9 : SUCCESS in 13m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/02f8b7058988466987ee13e5ff7c6406 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ef937d19d9f49b6b3d7c3ae0c8d26d2 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f609e220596f4cce93b810ed347b16ed : SUCCESS in 17m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6d096754f7c545c9b628b9d34b2fa319 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3a8e910c2f0b415b8e5c6d94a6d770ac : SUCCESS in 41m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cd14e6f35b4d4a0e874bb465d8453ae7 : FAILURE in 1h 20m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71c778d078e542f78add31115670b978 : FAILURE in 21m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c7dedf0c54574794af20c440a00add4c : SUCCESS in 47m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/85ce9d906c024464911a8ae5a747ec0d : SUCCESS in 45m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2540c3b58924480db6b28bcae40ba150 : SUCCESS in 36m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5bca8226c1c3462abd620899147f4afd : SUCCESS in 42m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cfc981bbe0e4178bbb9a65474bf66af : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f981f8ff2f254860a33b14842cd6c276 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7df02ca3bd14f12a1f4a526039f7b2f : FAILURE in 9m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa2c432f559a4ae99d7a60b40fe30138 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":2},{"id":"02a87a6ffda9860c8d1e0fafdc9313fef8256f83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-08 00:56:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"161706dd6d71e1878cdb30acab5796e2e05b956f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 02:00:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfe53deb339496d824e83951faf6928 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/02c588dd3e0e4970901a370d0ede4a36 : SUCCESS in 57m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6947c6e829d49d3b0744c6b19ee8a3e : SUCCESS in 47m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9110ca4939924246ade90177851ecbfb : SUCCESS in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8efc540cb9554c80bdc969aad919d4ec : SUCCESS in 9m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/acf12eb627e640a8aedbc1586a32871a : SUCCESS in 11m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/382c70a95c7f46c3a24d560b0c779bd9 : SUCCESS in 13m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/823f1faa7fa5431d9c1e3e366afc89a4 : SUCCESS in 15m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed6cbb5dd8a41a6b991f1ebc95f6007 : SUCCESS in 14m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1199899be7ce48008fbad83cff26c93f : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ded0e932e5dd47e29a4238c4c7103748 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5cab132f40f447df9d7e28f7d6ead029 : SUCCESS in 16m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/414dbd012a8543afb9bb6c137e6de0ee : SUCCESS in 22m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dc9ede00b67941ed9fc848d3d681ee38 : SUCCESS in 23m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/df9a67c65807401bbe946f6d9a29de39 : SUCCESS in 17m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e12801ea3db405993b6d915bc92a685 : FAILURE in 27m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9937dbc12ab4e97979ddaa62d5c7fee : SUCCESS in 45m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e5ea56e7ecb2494c8e3efe360616cd86 : SUCCESS in 55m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/465359e1aa5b4877b80d98d458483c85 : SUCCESS in 42m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48ef7f1f94ec4063a1bb6decb691fac9 : SUCCESS in 37m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1808f4ca7884486b97aa7f2d79b1c4b4 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfa5e6b0dcc42ef80a985a05166e250 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/219078dcb8a940f2aef67b0c92732e44 : FAILURE in 8m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9783ee93f25a45d6a65f3d3dc6345213 : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b5d44ccbe1d5d4dd4b6c9cfd21dd0456154a9dba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 02:00:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"875fb7417260d7b63abcf34b85b977593ce41f9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 04:52:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fbd492214842a393d104549a92b7a4 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/3c553b868a9747349e619bba985cb615 : SUCCESS in 1h 19m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a6b1882c185497abb4ec6ece2d126b4 : SUCCESS in 56m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bd9e54968c7a4ca39670f1ffae933bad : SUCCESS in 8m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/86fd521375e148e7abbf3c70a2e1dfed : SUCCESS in 9m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/db8f1d357e2a44a1ac7b8c1e92622c9d : SUCCESS in 13m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/98d3a004edea49dda0bef6c12cc92bb8 : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/947c5e750d794fb9920cd638429564de : SUCCESS in 18m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f30cf7b9765427db1ddbecfcbbf8283 : SUCCESS in 13m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b768882f4da41458275b4f1dfcc3681 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be708d9b142a4df9b28645781c7bdd25 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c739d3d5c6e2410abc369d98486e1719 : SUCCESS in 23m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/227675c084b24529b6253d91a19af554 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d11e3f25d5344c75acb78abb1b2d5ec2 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2d1359d774614cca9850e39461294588 : SUCCESS in 18m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c4d5748ab8ca4bcc978a0e8905af4ac3 : SUCCESS in 59m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a9163f02661a4b18b082fe96632dc44c : SUCCESS in 44m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf266ba498174ad59ad77049489d5cd7 : SUCCESS in 40m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5706a1d797ca49609ea07477de97534e : SUCCESS in 42m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/473d073ac63c4395bfa03e10adfb1d61 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bcd50d598af40e888ac53dc7d07ce10 : SUCCESS in 7m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ff3b3248845446fbce8a4313f23960f : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":2},{"id":"86549bed3484f1b17be45f9396c963f13fead835","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 04:52:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f330e6a360ed2fd691210d2b0e04cc23b410593a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 04:53:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f471b70383884c84aa3b3a4f7a810021 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/60503dc13789485b965920f556bf4d33 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6f3a32c2b07d445ba8f22470d3ddf50c : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87e88782aa14df5fe112fb0921aa9e4db74a153e","revisions":{"22878d8374f7b9ec8f7f5a090879011b9bb2231d":{"kind":"REWORK","_number":1,"created":"2020-12-17 23:19:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/767644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/767644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/767644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/767644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/767644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/767644/1"}}},"commit":{"parents":[{"commit":"28cef6767fe28062563d1bb18722184139df50e8","subject":"Merge \"Fail if swift binary are missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28cef6767fe28062563d1bb18722184139df50e8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-17 23:17:44.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-17 23:19:06.000000000","tz":-480},"subject":"s3api: Better-handle SHA mismatches during CompleteMultipartUpload","message":"s3api: Better-handle SHA mismatches during CompleteMultipartUpload\n\nAlso, allow upper-cased (and mixed-case) SHAs.\n\nChange-Id: I3d6e2e4542a5ed03a6d31ec0ef4837d1de30a045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22878d8374f7b9ec8f7f5a090879011b9bb2231d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22878d8374f7b9ec8f7f5a090879011b9bb2231d"}]},"branch":"refs/heads/master"},"87e88782aa14df5fe112fb0921aa9e4db74a153e":{"kind":"REWORK","_number":2,"created":"2021-01-07 19:14:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/767644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/767644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/767644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/767644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/767644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/767644/2"}}},"commit":{"parents":[{"commit":"8c611be87633c6bab14efe94d7d6d38c6f71859e","subject":"Merge \"Memcached client TLS support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c611be87633c6bab14efe94d7d6d38c6f71859e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-17 23:17:44.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-07 19:14:22.000000000","tz":-480},"subject":"s3api: Better-handle SHA mismatches during CompleteMultipartUpload","message":"s3api: Better-handle SHA mismatches during CompleteMultipartUpload\n\nAlso, allow upper-cased (and mixed-case) SHAs.\n\nChange-Id: I3d6e2e4542a5ed03a6d31ec0ef4837d1de30a045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e88782aa14df5fe112fb0921aa9e4db74a153e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87e88782aa14df5fe112fb0921aa9e4db74a153e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
