)]}'
{"id":"openstack%2Fswift~768668","triplet_id":"openstack%2Fswift~master~I62b12f69e9ac44ce27ffaed320c0a3563673a018","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-10-24 13:53:12.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-10-24 14:53:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62b12f69e9ac44ce27ffaed320c0a3563673a018","subject":"Clean up some replication error messages","status":"MERGED","created":"2020-12-29 06:08:00.000000000","updated":"2022-10-24 14:55:27.000000000","submitted":"2022-10-24 14:53:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768668","meta_rev_id":"f457c89a25f3893f2ca1d43ac90caea7fbcb6c23","_number":768668,"virtual_id_number":768668,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-24 14:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-10-24 13:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2022-10-24 13:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 07:33:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 16:42:58.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-10-21 14:39:34.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"4aa1db9611e165e8b633b5d222468fe3237ca676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-29 06:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b763d3732b7b3507144a769dd4c416e00db979a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 07:33:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2a7c949a4e41278d64d68b4cb87c6f : SUCCESS in 11m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/c9a4af1c16f24b6ca73ad8185ba3db40 : SUCCESS in 1h 06m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93ba5553a9d946dfb0c1909b2fbf66e5 : SUCCESS in 45m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08bf1e47b9b64fb5979ad48394470ced : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4a85bd6ed532459ca1b6197e570b351a : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/87e5507281a54c0380973073cd784f0d : SUCCESS in 13m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53c57865865e47ccb5210e72bc8acbb2 : SUCCESS in 14m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2bf7865791342d28985017e1bebb645 : SUCCESS in 13m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/64708409818742fd971ea47434173fc3 : SUCCESS in 14m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1136c29688cf430b9994e235794aa47d : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/554294ea68d14a33a4a6a86a1397f7ec : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5696cf4941494265a14fee4a0d67de63 : SUCCESS in 16m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/80da316879cd42a1a467eafb804e29bd : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ea5020c7e7824dd9ad2af2e9d6514bf8 : SUCCESS in 22m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8abd5c1a45b04f649ee36ed3da44b232 : SUCCESS in 23m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d36636d37fc14844a949baf59a985466 : SUCCESS in 21m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/acbdb7eb1c004030a5a2dff9586efd3b : SUCCESS in 46m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/abc55cc09d564f529fa237d1d6587293 : SUCCESS in 45m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00e4a63ddad74418a535e49c864f70de : SUCCESS in 56m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f137d28b8fad4849b8831c3137a50fa6 : SUCCESS in 42m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f94ef1f2bdb743c1a701106cfccf413f : SUCCESS in 14m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934609eeeca743b68e7894dfecd00301 : SUCCESS in 7m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55597a85a6634e5cb682eccbfd3e35ff : FAILURE in 12m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91f219219e804ccaa9f556e1969a05f2 : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"effb719da383b42843c7d1be0703621164bd8d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-05 21:30:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc20d1530f20d9a1bb1196666f473e109787cebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 22:49:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582292f3202e4692ae299ec1aca07f8e : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/ff51717dbd6b4d2a82b0207848b826d6 : SUCCESS in 59m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ff6d11c2887b4996b220fdb9682e8b58 : SUCCESS in 56m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48e1c6a6ec9e4dda8d32f32edeebdd23 : SUCCESS in 10m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/35aa6c0879984f76b15ac04346b99815 : SUCCESS in 11m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81b4a7918e484f268ed06294d0fa28a3 : SUCCESS in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3fac68e88ab2458ab28032a3fa212dbc : SUCCESS in 12m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/54619a591615474ba0d33fcbe31df3da : SUCCESS in 12m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/92dbd8442d624712a58072b523cf5ea9 : SUCCESS in 12m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35a98a53c5d54a6e9f99179c1f72eb3e : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/23b8ee6f57134f4abc6b7231995f1297 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a1cc4cf22b44f458cbbb512c614f6fd : SUCCESS in 17m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6d13387ddbb544488d86c78db056b62b : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/36588cbc9cd747c1aec5f70b770ad5e1 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f54649616ae64675be7708a48d858b18 : SUCCESS in 16m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bdfe3bf638194950892c4aad30996da4 : SUCCESS in 22m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ff5faacd31944f6685e87695096ba183 : SUCCESS in 40m 57s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/777e549704834ed2acaa632f46bb85b4 : SUCCESS in 43m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/038a20d2bb50405d850ffa1d1c7eba51 : SUCCESS in 41m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42334cbd41c84f7a9831cbbdb0a6ad24 : SUCCESS in 38m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f38319cfe174590aed354b89ba1d168 : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/035b76ab128843879395babb557321ab : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b7a5fc313a01419aae5933bba482e832 : FAILURE in 7m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8d347b995ea4cd7a0495e384374dc41 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1138768d7054e58793d82f1b4b5d5f6ca996b746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-03 20:10:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7791b2e293553ccc5e02d8df7545ac6ce5f2276e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 22:13:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93559948a1554117a21860a7ba4ecd81 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3cd7a7c1e0c24bd39d13f7f0537b38b6 : SUCCESS in 1h 01m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/78f08f7f3a8f4181b9b59b96af0c260d : SUCCESS in 45m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/875c0843e9ec4539b1f5dd470d114276 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fd3110eda7eb422b81521c7b44752dd0 : SUCCESS in 10m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1bd389822a5c4d629056bba5ace0694a : SUCCESS in 14m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a96414b9c649ec910a217a6e2dc1c4 : SUCCESS in 19m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/872ec83e499c4ba4ba586d8d70c884cd : SUCCESS in 38m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ae2d646eb7334d139267a609bbf3a3b2 : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c559bae959e04224b0a8d3bed832a2c4 : SUCCESS in 20m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1853b177d04e450ab376b7af633e46f3 : SUCCESS in 20m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b2a2ddff548490084ba502bff3c5ba3 : SUCCESS in 18m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3120eebaec2d4acd9067c025bb264ded : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e37721b6cf544b89ae6c75fd131fcfc6 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8ac4f91972274700a10acd3fbe8afefb : SUCCESS in 23m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/759a90d7262e4ec7afa635f45d4e469a : FAILURE in 22m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f199e3fb5e03431d9eb701c56c79def3 : SUCCESS in 49m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/954a2441ce27459f9993ba92e196b908 : SUCCESS in 49m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59a8568cfe704d56a8d0b9753515b53b : SUCCESS in 48m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ddd976b97ab84ad9bbed372e9308af04 : SUCCESS in 37m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaf20bc074bd440e92f21784b058cdc0 : SUCCESS in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095c7e05d72e4ebd8c227142f569d1d9 : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34b9bdc12a0b4dc993be85d1a05f97a8 : FAILURE in 7m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/333351fe456d40319ef5715b2b470494 : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":3},{"id":"55fc8bdb6cf71d0f5d7fc7661a91628bb0daf502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-20 16:35:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"71fafc1a311b89563421765a23692c7bac1a4daa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:23:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54b02fa239246bf9290b28b2b40dfce\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/921e1f8b73dd4b11aea78948867d5dc3 : SUCCESS in 37m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/280444dc79b943739f2d5b16533bbe8f : SUCCESS in 36m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cca434116f5427f87267fb967422f6d : SUCCESS in 36m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d469dff63a74ba7b25bf36d547bae88 : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":4},{"id":"4a76d7ae212ed63efd26485d5f3bd5105f035d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:42:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fac8829419b74fda9634e6f6e3c1fe40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9992fc73ab4d53b2724ac25f366135 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/e41e5fa20dac469088aad6863addfc70 : SUCCESS in 40m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cab0276f6bfb4c28886763bd5792a837 : SUCCESS in 39m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/237a7fbbea374292b165c69b5f47e3a0 : SUCCESS in 56m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/025328db2d5c4c6f9385a9e79a3f2b64 : FAILURE in 17m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85a25900741640dfac96b05555f16639 : FAILURE in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0eeedd74fdf04a269cb7552d066022dd : FAILURE in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/acd2d04cc13444089a908ecb28dc5c8b : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8081873bf49541ae9ae7baff364008f9 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/74535e78454f4572be5942bbb323072e : SUCCESS in 21m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef28f4c853364f468fe1ddb36c0661fb : SUCCESS in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de5e2fece49f4dff9ce225a254e8a9e9 : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cf9e2d4d6dc943b9a0f2dc97aee81692 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9837850dfdce4eb6be6d234330832873 : SUCCESS in 23m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a38fd7035df94372ba14cc218c5fda34 : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a355523901834cdc87e806c3e0373a3d : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/acddc58c61b54547b5d6cd5275a9fdfd : SUCCESS in 12m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc4e8b3006bb4c13b0ee0ab4890537f8 : SUCCESS in 20m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ce7a9fc36ec4cf49d350d86d6aa2ab9 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff91f4da2b6f43628f892dff5d8655ed : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9267d2a16c546bea528aa921b98e9e7 : SUCCESS in 6m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/262d5642f29343a8bae28daa5546541a : SUCCESS in 1h 00m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/62fcbf789d1c426f9dc5991b6d41a5f1 : SUCCESS in 1h 04m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/834a8787e5004db9980a5b2ca61f6933 : SUCCESS in 41m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9811ebfd750482c876a09dcf0ba4eaa : SUCCESS in 41m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9dc03f5427c44ccb65d257ad9655354 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2a39ac5bc84b8d836ea512588f403e : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d6835a80b0a490db13107a59b8befb2 : SUCCESS in 21m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23ce84ae235e4ba3a55c69dac85f1278 : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":4},{"id":"971abad7f532a10c24bbfc26f143103aaf924afa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-21 14:39:34.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5fff811cfeba47c1c087f277ede6cff419d7c197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-23 22:24:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ba12a286aacd9548f8c7586f4cf58f063bbb4650","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-23 22:24:54.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d9b91b2cea4a7ff56fc205d92a960a4d06cab678","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 22:59:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b89c5dc773974e3788444ef7702f4bd6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/023a5fb1809840ca8d2eff674ccfc8d8 : SUCCESS in 28m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d63547305b243ccb2e450becf42c947 : SUCCESS in 28m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/68c942efefcc47e196c89c3ebdc66c2a : SUCCESS in 28m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d88bc0197a34493aa25f714390bc94b : SUCCESS in 27m 40s","accounts_in_message":[],"_revision_number":5},{"id":"c69a2d99cd4b3209ab668ffa81ec8d5c594ec495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 23:30:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89cb8d03003345c2a82f0bba5b09dd57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f56e978012945b18c8c18c4c8802bc6 : SUCCESS in 7m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/5bfdf9f3ac664cd9a8d2369a496ee802 : SUCCESS in 54m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84c89b0e4fd34e75a36fd02f192ccaff : SUCCESS in 35m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaf0320061ad4b70a61f0e3e7958c760 : SUCCESS in 45m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcd849332ea94731b53134b30bc60865 : FAILURE in 24m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d24d0237d64f42bdb3c922a9cf2978f7 : FAILURE in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8f7780e4c4b4a5ebe6cbc03ce00a81d : FAILURE in 17m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32c098c0fe77423583290728090ed285 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b19a4a7cf7f5457e9b293993691f436a : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1f1a7ab69544c8b9fc32f76ce986472 : SUCCESS in 14m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00815fe7be274780afaf62bf7ae3b959 : SUCCESS in 20m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6af9341aeaa4173b4778702419d1f1c : SUCCESS in 13m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e947d8e91a5406c893dfd28c96fd922 : SUCCESS in 20m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ade2c45e6f9245ef997c815b7e7fa77a : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d2f7ac344e234f0c8059dd05443005fd : SUCCESS in 21m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/86f6f6e566da455c84bfdac5b4a6085d : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92df7b90e48844afa3e3233a413de272 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2d26bca5127a4b70baa31457faa0da3f : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57751eed99f044a985fbdbe649be3dad : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/be82453a2b1b4b41a336c4cc01f2ee9e : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cd0b64fdd944cf49961eb4df67aa47f : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/186a8883f4104c5ba1f320f657a39c60 : SUCCESS in 46m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c06b3f221c14d0eaf434ca148ae2e3b : SUCCESS in 1h 04m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c82fe54b1f024f5c8fb694f44b17a60f : SUCCESS in 41m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fdfeb7bb8744d2cb62bf69a53095261 : SUCCESS in 33m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c5a48c292c042ecba58be35a9e91098 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d886f309e9d4895ad391d273469e136 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e55cc701ae34e418de88981dfa4e4a0 : SUCCESS in 18m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/006638a0c58f49698727d57ffd1bc133 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":5},{"id":"e9e2f4b80b8b7303affd21fc20a212076b1b955e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-24 13:53:12.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c8589eed5eb9749185833de6760bc61e44c12e08","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-10-24 13:54:13.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"de110e7dbff45520e3c41fd72fffb4346812916c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 13:54:48.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1135f57ef403e155e6f64fa80649ecae207daa5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 14:53:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd5a467ec76f432d8e1c2ece72d2963d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1674f531e2bf464c9e6911107693025c : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/4ba7e30f23f74a57ae6611ca359bdd84 : SUCCESS in 49m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9398fefb770545378664a1013db2fadc : SUCCESS in 39m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb8007f2b9a745b8bd54c6cfe45bb5bc : SUCCESS in 45m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1fddbd224ece4f9f97dbc82b5ec212b5 : SUCCESS in 8m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fe808851196f4b768556942119f11cf3 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c93bf9038e18449b8fb549e50f534407 : SUCCESS in 22m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35d32abe4ef048a29faaca2aa8a07f0e : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e54be4e95064c238b5ce6ba67dbcf36 : SUCCESS in 18m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b02d85012bf14a80b385fd5e3a2a1e47 : SUCCESS in 22m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e2f521a394c469dbc3b58cfdeca0ec2 : SUCCESS in 20m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ecf0c60ebc034437bcec8551f6a6fbfd : SUCCESS in 18m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/17c0eaec2a6543d2ace650d8a7c34e5b : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ecf4c5769cb3452b829ed6284ed55598 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c795c51e3a134c2cb2c14ee2f42e59cb : SUCCESS in 19m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ffed99c55af409083f042cf358f84c9 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1be2487f7e5747799a5c1509c6f89906 : SUCCESS in 7m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b65586ab9c3f407bb7bcb51773f765ab : SUCCESS in 50m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/28a77b3a526f4c2bb88733a4ff4a2f73 : SUCCESS in 58m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4313511658941e982abb60f3508e150 : SUCCESS in 35m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66dcc9ae52a0493fb9c2469c0e9b5708 : SUCCESS in 37m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80553d38dca449d9b7ceb67347048235 : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b9ec6bf9ba45df97da81f1e4aff0b5 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/806e28f91f4d435db0417f430bbc4589 : SUCCESS in 19m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bff40aecf5b248908642761ff4f13022 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":5},{"id":"f86fd3dff4a36797d3a844fca596de823bce9906","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 14:53:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f457c89a25f3893f2ca1d43ac90caea7fbcb6c23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 14:55:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/737f5e5f97d142e6a3b9e18c92f26240\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f9acfa823a54277bf3a2c28932526d8 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4deac9ed839a4f268b291ba6c73721cc : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/21a9afc6ff9048b2b6a8639a4d949962 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a7af17394ba62e3351fa73b8d93cc28c493e4f3a","revisions":{"18f6df944e25ddd986d8e4f0e47ee1e748c12dd3":{"kind":"REWORK","_number":1,"created":"2020-12-29 06:08:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/768668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/768668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/768668/1"}}},"commit":{"parents":[{"commit":"fb91993b47aa02209ba39382620789b14e10d6d2","subject":"s3-tests: Use my s3compat fork instead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb91993b47aa02209ba39382620789b14e10d6d2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-29 05:58:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-29 06:07:16.000000000","tz":-480},"subject":"Clean up some replication error messages","message":"Clean up some replication error messages\n\nLines like `Invalid response 500 from ::1` aren\u0027t terribly useful in an\nall-in-one, while lines like\n\n   Error syncing with node: {\u0027device\u0027: \u0027d5\u0027, \u0027id\u0027: 3, \u0027ip\u0027: \u0027::1\u0027,\n   \u0027meta\u0027: \u0027\u0027, \u0027port\u0027: 6200, \u0027region\u0027: 1, \u0027replication_ip\u0027: \u0027::1\u0027,\n   \u0027replication_port\u0027: 6200, \u0027weight\u0027: 8000.0, \u0027zone\u0027: 1, \u0027index\u0027: 0}:\n   Timeout (60s)\n\nare needlessly verbose.\n\nWhile we\u0027re at it, introduce a node_to_string() helper, and use it in a\nbunch of places.\n\nChange-Id: I62b12f69e9ac44ce27ffaed320c0a3563673a018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18f6df944e25ddd986d8e4f0e47ee1e748c12dd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18f6df944e25ddd986d8e4f0e47ee1e748c12dd3"}]},"branch":"refs/heads/master"},"55996f77959650dc2ded10dbdfb8bef390ba0604":{"kind":"REWORK","_number":2,"created":"2021-01-05 21:30:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/768668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/768668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/768668/2"}}},"commit":{"parents":[{"commit":"92b15b0436b2246e658af94bdc312bdffbebfc74","subject":"Merge \"Let developers/operators add watchers to object audit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92b15b0436b2246e658af94bdc312bdffbebfc74"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-29 05:58:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-05 21:28:33.000000000","tz":-480},"subject":"Clean up some replication error messages","message":"Clean up some replication error messages\n\nLines like `Invalid response 500 from ::1` aren\u0027t terribly useful in an\nall-in-one, while lines like\n\n   Error syncing with node: {\u0027device\u0027: \u0027d5\u0027, \u0027id\u0027: 3, \u0027ip\u0027: \u0027::1\u0027,\n   \u0027meta\u0027: \u0027\u0027, \u0027port\u0027: 6200, \u0027region\u0027: 1, \u0027replication_ip\u0027: \u0027::1\u0027,\n   \u0027replication_port\u0027: 6200, \u0027weight\u0027: 8000.0, \u0027zone\u0027: 1, \u0027index\u0027: 0}:\n   Timeout (60s)\n\nare needlessly verbose.\n\nWhile we\u0027re at it, introduce a node_to_string() helper, and use it in a\nbunch of places.\n\nChange-Id: I62b12f69e9ac44ce27ffaed320c0a3563673a018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55996f77959650dc2ded10dbdfb8bef390ba0604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55996f77959650dc2ded10dbdfb8bef390ba0604"}]},"branch":"refs/heads/master"},"485512dc9063d50480ed7344c4b5ad4bd6c8752f":{"kind":"REWORK","_number":3,"created":"2021-02-03 20:10:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/768668/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/768668/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/768668/3"}}},"commit":{"parents":[{"commit":"1f9b879547a5fd8fe1bbdbf42e20d7f73ac5d393","subject":"Run flake8 on bin/ files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f9b879547a5fd8fe1bbdbf42e20d7f73ac5d393"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-29 05:58:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-03 20:08:40.000000000","tz":-480},"subject":"Clean up some replication error messages","message":"Clean up some replication error messages\n\nLines like `Invalid response 500 from ::1` aren\u0027t terribly useful in an\nall-in-one, while lines like\n\n   Error syncing with node: {\u0027device\u0027: \u0027d5\u0027, \u0027id\u0027: 3, \u0027ip\u0027: \u0027::1\u0027,\n   \u0027meta\u0027: \u0027\u0027, \u0027port\u0027: 6200, \u0027region\u0027: 1, \u0027replication_ip\u0027: \u0027::1\u0027,\n   \u0027replication_port\u0027: 6200, \u0027weight\u0027: 8000.0, \u0027zone\u0027: 1, \u0027index\u0027: 0}:\n   Timeout (60s)\n\nare needlessly verbose.\n\nWhile we\u0027re at it, introduce a node_to_string() helper, and use it in a\nbunch of places.\n\nChange-Id: I62b12f69e9ac44ce27ffaed320c0a3563673a018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/485512dc9063d50480ed7344c4b5ad4bd6c8752f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/485512dc9063d50480ed7344c4b5ad4bd6c8752f"}]},"branch":"refs/heads/master"},"72df7b95eda34b7b3d82b74df6831fdcdf0fd4c2":{"kind":"REWORK","_number":4,"created":"2022-10-20 16:35:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/768668/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/768668/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/768668/4"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-29 05:58:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-20 16:34:59.000000000","tz":-420},"subject":"Clean up some replication error messages","message":"Clean up some replication error messages\n\nLines like `Invalid response 500 from ::1` aren\u0027t terribly useful in an\nall-in-one, while lines like\n\n   Error syncing with node: {\u0027device\u0027: \u0027d5\u0027, \u0027id\u0027: 3, \u0027ip\u0027: \u0027::1\u0027,\n   \u0027meta\u0027: \u0027\u0027, \u0027port\u0027: 6200, \u0027region\u0027: 1, \u0027replication_ip\u0027: \u0027::1\u0027,\n   \u0027replication_port\u0027: 6200, \u0027weight\u0027: 8000.0, \u0027zone\u0027: 1, \u0027index\u0027: 0}:\n   Timeout (60s)\n\nare needlessly verbose.\n\nWhile we\u0027re at it, introduce a node_to_string() helper, and use it in a\nbunch of places.\n\nChange-Id: I62b12f69e9ac44ce27ffaed320c0a3563673a018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72df7b95eda34b7b3d82b74df6831fdcdf0fd4c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72df7b95eda34b7b3d82b74df6831fdcdf0fd4c2"}]},"branch":"refs/heads/master"},"a7af17394ba62e3351fa73b8d93cc28c493e4f3a":{"kind":"REWORK","_number":5,"created":"2022-10-23 22:24:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/768668/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/768668/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/768668/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/768668/5"}}},"commit":{"parents":[{"commit":"4ed2b89cb78f06cbd08b8a3f94745613def12e97","subject":"Sharder: warn when sharding appears to have stalled.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ed2b89cb78f06cbd08b8a3f94745613def12e97"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-29 05:58:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-23 22:24:23.000000000","tz":-420},"subject":"Clean up some replication error messages","message":"Clean up some replication error messages\n\nLines like `Invalid response 500 from ::1` aren\u0027t terribly useful in an\nall-in-one, while lines like\n\n   Error syncing with node: {\u0027device\u0027: \u0027d5\u0027, \u0027id\u0027: 3, \u0027ip\u0027: \u0027::1\u0027,\n   \u0027meta\u0027: \u0027\u0027, \u0027port\u0027: 6200, \u0027region\u0027: 1, \u0027replication_ip\u0027: \u0027::1\u0027,\n   \u0027replication_port\u0027: 6200, \u0027weight\u0027: 8000.0, \u0027zone\u0027: 1, \u0027index\u0027: 0}:\n   Timeout (60s)\n\nare needlessly verbose.\n\nWhile we\u0027re at it, introduce a node_to_string() helper, and use it in a\nbunch of places.\n\nChange-Id: I62b12f69e9ac44ce27ffaed320c0a3563673a018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7af17394ba62e3351fa73b8d93cc28c493e4f3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7af17394ba62e3351fa73b8d93cc28c493e4f3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[]}
