)]}'
{"id":"openstack%2Fswift~769663","triplet_id":"openstack%2Fswift~master~If380410c17ed9ebc014b8198af0ea8d502deacc8","project":"openstack/swift","branch":"master","topic":"cache_shard_listing_doc","hashtags":[],"change_id":"If380410c17ed9ebc014b8198af0ea8d502deacc8","subject":"Document memcache max size with shard ranges","status":"MERGED","created":"2021-01-07 06:11:11.000000000","updated":"2021-01-27 21:10:47.000000000","submitted":"2021-01-27 21:08:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"769663-cache_shard_listing_doc","meta_rev_id":"ddee38a7a67526b86f849e4e7f2df452469db818","_number":769663,"virtual_id_number":769663,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 21:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2021-01-27 16:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2021-01-27 16:05:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 07:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 09:37:40.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-01-08 05:37:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-01-27 16:05:34.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"77bdf9fb5868e6aac162e9e27435abe53b39302b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-01-07 06:11:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bd08131ce95e0dd2a41bcf1caa2eed8584837e8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-01-07 06:17:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03f6c8ea60cedfff5d495ae6d724e995fe887d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 07:10:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04d0dd4869c4d7e9f0160180490a638 : RETRY_LIMIT in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e4de6edc0df4d72b7c0253932de6cfff424d279d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-07 09:37:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nthanks Matt!","accounts_in_message":[],"_revision_number":1},{"id":"6baf36033b014f75663e8c52d96c5174aec2cb09","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-07 09:42:39.000000000","message":"Patch Set 1:\n\ncould you wrap lines to 80 columns? thanks","accounts_in_message":[],"_revision_number":1},{"id":"85a611343874f5722465a007a7146d4b2cb86154","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-07 11:56:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa67c064a5882b8143d626b249613b7c47e6fcf7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-01-08 04:25:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd9140d0a82b7585fe486b551ac7db83250aa9e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-01-08 04:27:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2f9d31d30e75e47c27d913abe34f55222558aa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 04:38:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1856f4d4afb64eac8e24d606b255daff : SUCCESS in 9m 21s","accounts_in_message":[],"_revision_number":2},{"id":"88ce9b5e5899e2e9c8f6c12e0a24c0d82f09ea35","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-08 05:37:35.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"88adb637d067504cd82f8a1e573dc489a4534434","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-27 16:05:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)\n\nthese large cache values are going to be an operational issue for more deployments once auto sharding is rolled out - I think having this section in the docs is a win - obviously it can be improved as we learn more; but I don\u0027t see anything that\u0027s wrong/misleading.\n\nI think having https://review.opendev.org/c/openstack/swift/+/769868 will help a good deal - after they both merge we should add a note about that log message to this section!","accounts_in_message":[],"_revision_number":2},{"id":"e2bc5c4a466b6e1ba3bc0b7d720eaddab4f990ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 16:06:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bb88c82b9bd602d902344acee8d37a58af3959b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 21:08:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31ea15dca8942debb8808d3af383d50 : SUCCESS in 8m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4dc06c1aa40943d989f7b738713104e5 : SUCCESS in 9m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9944825caefe45228c36af2fa5c09d06 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ac3256bbfe21d5fea5e422db89deec73807509c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 21:08:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ddee38a7a67526b86f849e4e7f2df452469db818","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 21:10:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2882010f6fcf4371be99ec30d3b582f6 : SUCCESS in 1m 21s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/34bd1d8e7b2b4665bb9b89c1e2b40667 : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/77479dbb96064abeb3c66df749083082 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8df3d1505a4e034983c609cb6f8f5e9a0fc945f2","revisions":{"23261b1da5f6ade02e37ec4ab5fd1d9445b8e3b8":{"kind":"REWORK","_number":1,"created":"2021-01-07 06:11:11.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/769663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/769663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/769663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/769663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/769663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/769663/1"}}},"commit":{"parents":[{"commit":"077ba77ea6f735c98709b60e1fc9aa9f0f90676a","subject":"Use cached shard ranges for container GETs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/077ba77ea6f735c98709b60e1fc9aa9f0f90676a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-01-07 06:06:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-01-07 06:06:55.000000000","tz":660},"subject":"Document memcache max size with shard ranges","message":"Document memcache max size with shard ranges\n\nWhen you start getting more then 3k shards in a root container the\ncached shard range listing can get bigger then the default max size for\nmemcache (1MB).\n\nSo add a mention about it in the configuration guide.\n\nFixes bug 1890643\n\nChange-Id: If380410c17ed9ebc014b8198af0ea8d502deacc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23261b1da5f6ade02e37ec4ab5fd1d9445b8e3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23261b1da5f6ade02e37ec4ab5fd1d9445b8e3b8"}]},"branch":"refs/heads/master"},"8df3d1505a4e034983c609cb6f8f5e9a0fc945f2":{"kind":"REWORK","_number":2,"created":"2021-01-08 04:27:49.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/769663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/769663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/769663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/769663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/769663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/769663/2"}}},"commit":{"parents":[{"commit":"077ba77ea6f735c98709b60e1fc9aa9f0f90676a","subject":"Use cached shard ranges for container GETs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/077ba77ea6f735c98709b60e1fc9aa9f0f90676a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-01-07 06:06:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-01-08 04:24:28.000000000","tz":660},"subject":"Document memcache max size with shard ranges","message":"Document memcache max size with shard ranges\n\nWhen you start getting more then 3k shards in a root container the\ncached shard range listing can get bigger then the default max size for\nmemcache (1MB).\n\nSo add a mention about it in the configuration guide.\n\nFixes bug 1890643\n\nChange-Id: If380410c17ed9ebc014b8198af0ea8d502deacc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8df3d1505a4e034983c609cb6f8f5e9a0fc945f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8df3d1505a4e034983c609cb6f8f5e9a0fc945f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
