)]}'
{"id":"openstack%2Fswift~770124","triplet_id":"openstack%2Fswift~master~If75803648169f85b854c3d5d8784aaebbd93805b","project":"openstack/swift","branch":"master","topic":"p-ssync-non-durable","hashtags":[],"change_id":"If75803648169f85b854c3d5d8784aaebbd93805b","subject":"Refactor reconstructor probe tests","status":"MERGED","created":"2021-01-11 14:02:10.000000000","updated":"2021-01-11 19:06:10.000000000","submitted":"2021-01-11 19:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770124-p-ssync-non-durable","meta_rev_id":"19efe3c060d955e1d1d156339d02f83647cb35ce","_number":770124,"virtual_id_number":770124,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 19:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-01-11 14:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-01-11 14:54:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 14:54:53.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-01-11 17:25:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fa223a0b770b34270377c0f0dcbb99269d6f69c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-11 14:02:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70679650e04c14971af564bdf10bcc5ae1fccd86","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-11 14:02:41.000000000","message":"Patch Set 1:\n\nThis in in preparation for https://review.opendev.org/c/openstack/swift/+/770047","accounts_in_message":[],"_revision_number":1},{"id":"a1415a528293f72ed1e2a0f21fbd375c5c89dd73","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-11 14:54:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nseems reasonable!","accounts_in_message":[],"_revision_number":1},{"id":"90c59b158dad4303f3a82dee42162d608b66dc9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 17:25:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab11795980e4609a90bca8b4606746f : SUCCESS in 10m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ec751aa4b134d26b571189309e88a16 : SUCCESS in 44m 32s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/113223cfb5de4901a5b58d37fa1ff201 : SUCCESS in 46m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7a76c560a44845b54a3da57dc9643f : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"20a07170e93f7528e89b5dd6b31aea530dc4bdaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 17:25:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b587d72427c22afa2e7b35f7e7ddd104831154c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 19:03:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e42bcf9c94454f9cee9b6f7204f0cc : SUCCESS in 10m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3f860224069e4a14841b142ee442c0ac : SUCCESS in 10m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c737a25c58ff497dbae504c468c00046 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78ef124b18bd46a486fe689371fe5509 : SUCCESS in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e90d81a695ac4dc69037220421c2fdd8 : SUCCESS in 14m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d2463afd48f456ca220bcf8b5be32fb : SUCCESS in 14m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ffc9fc78f36b48afae84f215017427b7 : SUCCESS in 12m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24413cf0d3cc412bb9ab729f0dc0ce88 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d5e51b42e6448f5b89e46bc351ea29f : SUCCESS in 15m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8e148a59b714eccaf8367d7fd6408fc : SUCCESS in 16m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/159c427179ad4d11a0860766c581b7af : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/88445e61f9484661b7f573682d00dff1 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c1de928a2a84003aeddcf84083d36f1 : SUCCESS in 16m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e0d604b8ae0840d0925ee3ce577782dd : SUCCESS in 42m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7546aa8eb580476c9c52018397ca9db9 : SUCCESS in 41m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7b8361420e465d8d8f328620b03e75 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"66d76243582663ef16175ccc34a8143c1678aa2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 19:03:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19efe3c060d955e1d1d156339d02f83647cb35ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 19:06:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae2f2053c104a44b0a835cf1adafcb5 : SUCCESS in 2m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/71d97709058e4dabbc43b2df8a7e808a : SUCCESS in 1m 07s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b50488da6f3f457da8471f6534040d32 : SUCCESS in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"128f1995088d11e6d41902271b683750fb2e770e","revisions":{"128f1995088d11e6d41902271b683750fb2e770e":{"kind":"REWORK","_number":1,"created":"2021-01-11 14:02:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/24/770124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/770124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/770124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/770124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/770124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/770124/1"}}},"commit":{"parents":[{"commit":"73cc68f55b78943f62f72d500a81d577c30e464a","subject":"Merge \"gate: Add centos8/py36 experimental jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73cc68f55b78943f62f72d500a81d577c30e464a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-11 13:51:03.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-11 13:57:55.000000000","tz":0},"subject":"Refactor reconstructor probe tests","message":"Refactor reconstructor probe tests\n\nRefactor the reconstructor probe test to share common setup and helper\nmethods.\n\nChange-Id: If75803648169f85b854c3d5d8784aaebbd93805b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128f1995088d11e6d41902271b683750fb2e770e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128f1995088d11e6d41902271b683750fb2e770e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
