)]}'
{"id":"openstack%2Fswift~770796","triplet_id":"openstack%2Fswift~master~Ib96d402b9fcb4245e0e79e273643332559f16424","project":"openstack/swift","branch":"master","topic":"p-fix-test-verify-server","hashtags":[],"change_id":"Ib96d402b9fcb4245e0e79e273643332559f16424","subject":"Make test_verify_server independent of test env","status":"MERGED","created":"2021-01-14 11:58:34.000000000","updated":"2021-01-15 08:23:31.000000000","submitted":"2021-01-15 08:20:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770796-p-fix-test-verify-server","meta_rev_id":"4fd6df5c42172f8eed91558697573899bfbbbfcd","_number":770796,"virtual_id_number":770796,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 08:20:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2021-01-14 15:38:32.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2021-01-14 16:24:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2021-01-14 16:24:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 15:38:32.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2021-01-14 16:24:24.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-01-15 06:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6cede894ce615690aba5993c6d32d20a5bcfdbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-14 11:58:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcb554d3409fef4cc2fe6e032439ecbc0a954ccc","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-01-14 15:38:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b38f403610958ecf6835c37c3240835e5f9b15a4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-14 16:24:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e57927a17a16ccf9a570e02d789e9c320b48c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 06:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f81201884e4f8c84b0bde3806c33bb : SUCCESS in 9m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/725af4dc833641a8834b6a56a1d805fd : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa8335984c8e4b7f9de1b5dac24070b9 : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0bb881736b294a3e8ed4d8be2c08328a : SUCCESS in 19m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81657ee3f9024879b413b052940fbdfe : SUCCESS in 13m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b81706252bf54de1bebcb9746da9f08a : SUCCESS in 14m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/14e0d9bdeace44e28c9d4f3197e5d79f : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ed54e5be65c456da5de3299c7d309d2 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9fd991c1647e4279b5b5374ccff8323d : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/341ff1983f3e48ebbd594cb4ab33d0b0 : SUCCESS in 19m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dfd0696d45a5450c928b9022e5cb6079 : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c1e913b95b91465fbb803be2a10925cd : SUCCESS in 25m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/efc6cb8d60cc418f97975a96ae957e11 : SUCCESS in 17m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed9a48baa54c435d89c5f5046e349ef8 : SUCCESS in 40m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1758fdc4c82144fbbc620a9a31fe7911 : SUCCESS in 43m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c9e417d5f48400e885674a3b1a7b4e9 : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9afb881f634ea892f837e6db23bced : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4186b68874647378a0e2efed4f57b46 : FAILURE in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a898554b42e0837fe291325189e7124151f3fa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 06:05:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c723c9def5c561e17c6fccfd3c81ba7738b237b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:20:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98cdf9f1e56453496100c1a53d96f2e : SUCCESS in 8m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/44da614be0de402a81b26bdf5c9500a2 : SUCCESS in 10m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b2751b4863b7478b9cda30c02dd710e2 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd290b6cb8284d27a55dee800cbfeea1 : SUCCESS in 13m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/126eb611788a428da33339d607654dd7 : SUCCESS in 14m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c8c2c002b65c4aacb3b17490581470bd : SUCCESS in 14m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e7e4c27e5a5a4357872d188a6bde3a4b : SUCCESS in 14m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d6346aad092493c8907ffd00d0d3003 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5d77df4fe944b5eb0d034dea3e55308 : SUCCESS in 24m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5575b1c08b2b48cfb9f9c6514a4ae85b : SUCCESS in 16m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ec420667622c4cdf8bb501c7a8fa1b9c : SUCCESS in 20m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/df22da540739461584d209eae8a29e61 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c9380bf688ab46fb8cd4076ed3375b30 : SUCCESS in 20m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9048a214c22d442f90552b0593acc0c5 : SUCCESS in 38m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f310822a81641de97617e13ccdf318b : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa9630e4d79145b78c1989033ae2d208 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1279da4b6e47b78ff02a95df7e9470 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"14287c9f18d557bd7fc9094ae59ca2168f0c726a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:20:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4fd6df5c42172f8eed91558697573899bfbbbfcd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:23:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b9a2c2146c415f8f2006f1d8f54673 : SUCCESS in 1m 36s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/475a183c25904c3c845c3c94148174b8 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/07acc5b0e756482da8d80e90431042b6 : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cf4edfae5c564236e2cd952a41481e278b8979f","revisions":{"7cf4edfae5c564236e2cd952a41481e278b8979f":{"kind":"REWORK","_number":1,"created":"2021-01-14 11:58:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/96/770796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/770796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/770796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/770796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/770796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/770796/1"}}},"commit":{"parents":[{"commit":"3cb58f8efc76f5d49c3836c4288d5a05a3678ef4","subject":"Merge \"unit test S3 SigV4Request X-Amz-Content-SHA256 handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cb58f8efc76f5d49c3836c4288d5a05a3678ef4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-14 11:56:01.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-14 11:56:01.000000000","tz":0},"subject":"Make test_verify_server independent of test env","message":"Make test_verify_server independent of test env\n\nPreviously the test would fail on a case-insensitive filesystem or if\nthe test environment happened to have a file such as\n\u0027swift-test-server\u0027 on its PATH.\n\nChange-Id: Ib96d402b9fcb4245e0e79e273643332559f16424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cf4edfae5c564236e2cd952a41481e278b8979f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cf4edfae5c564236e2cd952a41481e278b8979f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
