)]}'
{"id":"openstack%2Fswift~771742","triplet_id":"openstack%2Fswift~master~Iabb53f5e4081d580d016bbf75d86e1d75e1f20bb","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iabb53f5e4081d580d016bbf75d86e1d75e1f20bb","subject":"Adjust initial month value from int to str","status":"MERGED","created":"2021-01-21 01:39:26.000000000","updated":"2021-01-27 02:53:37.000000000","submitted":"2021-01-27 02:52:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"771742","meta_rev_id":"2b0fdfeb6b354a1c34371b35e10cbc6f03f0085b","_number":771742,"virtual_id_number":771742,"owner":{"_account_id":11534,"name":"david","email":"david.cole@sohonet.com","username":"david-cole"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 02:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-23 05:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-23 05:34:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 03:08:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 19:15:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f5e3e311a4c5c7c15e55113db2068053b7cc6393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11534,"name":"david","email":"david.cole@sohonet.com","username":"david-cole"},"date":"2021-01-21 01:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a93a3a1747284a245a0ef5e02281023e7cd0c307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 03:08:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4318bab2513d470db8314d04a43b5ce4 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/fccc8ad20f7141c9954dce097ac3baf7 : SUCCESS in 47m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82ef51ea686949928c2343836da38b96 : SUCCESS in 55m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50820921738e4673897aec62bc605db4 : SUCCESS in 9m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/baa35bce852c4c08ace6fff5cb207350 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3caba7d73b024fd9b1771612764255a5 : SUCCESS in 14m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/863ed5f9aef94484a4a0618eeba26f0d : SUCCESS in 12m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8f97ea181b9a4e4698b245e5b28b22b5 : SUCCESS in 14m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9b525546f2c147d9b8b37e583933b18e : SUCCESS in 25m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92c39765a33e46e2abda8d522630fcdd : SUCCESS in 29m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d75a1bb103464ddfbc3f553188d22b30 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bed3a21fd3c446f29b44830af4023432 : SUCCESS in 29m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fb51513c228548b68b3d1f8df48f3288 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b8b73dee7122489395362c1c5bdd903d : POST_FAILURE in 19m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ef2e655893374dfca03df0618ec352a5 : SUCCESS in 17m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e674fe4b5e86489894c9af30295d28a0 : SUCCESS in 24m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/288f082bf36d42c18bfea58627fe6649 : SUCCESS in 42m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bed691bf8ef44f06bef771f281d43bc8 : SUCCESS in 45m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acdf89e1a9234e9694e22c9d56dcf5be : SUCCESS in 37m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72d80ba2d5774efd9b02cb089581bfa7 : SUCCESS in 42m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50f9c770730146debfb2b1df10ecf20b : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e7843bc243408480163e6dbd107b05 : SUCCESS in 8m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a67212f26b647ce9d925639b2ded1cb : FAILURE in 9m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb0f137d264d4c90ac3596bb6f9b6e53 : SUCCESS in 1h 26m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4c68f16ffc51ab4ad7bb6469e32b03e446bbed37","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-21 19:15:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThis has me thinking: surely the kern.log timestamp format is configurable, yeah? I wonder how tolerant we should be of other formats... 🤔\n\nI\u0027m concerned that while this will fix running it when there are no January log lines, (I\u0027m pretty sure) it breaks it when there *are*.","accounts_in_message":[],"_revision_number":1},{"id":"fd723f48a90f3261f5de181c7e26556ff0f63f29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11534,"name":"david","email":"david.cole@sohonet.com","username":"david-cole"},"date":"2021-01-23 01:15:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d9da0cc7b1ea7d9d0d419d6bbe38f4d0273a812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 02:29:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd26aca37ef4e46ae5c03a5194ec2bb : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/cfcbd5ba4c58423c86d40e1c46eced35 : SUCCESS in 1h 07m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e281efd86a443608026e8d864128a32 : SUCCESS in 48m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e91a7bef58d74823a1494fd855d1c1a8 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/28bd2c934cfe473dab68bf53d2d8671a : SUCCESS in 8m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79057e84674f4d25bcd868dab85fd8a2 : SUCCESS in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90c9501c7332428bba825d17b266ae60 : SUCCESS in 13m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/320b3270ae7f4ad7856baee3463214b9 : SUCCESS in 12m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4f52d23746a041399a6e672fbef5cf49 : SUCCESS in 13m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a136054592034452a5a63102691b003d : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8947a730ca5a4fee8a7aa256d100271f : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7f6a0cc4f89e479bad6efa10c1cb4db1 : SUCCESS in 17m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6fc280b7de834cf5a86552941419b958 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3368ef54174f4baa8c9d37e98d724e8d : SUCCESS in 19m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8d772b4edaa74d7fb01278d9f18eece2 : SUCCESS in 16m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46f2bd3570234d83a75d6a40b7c008a2 : FAILURE in 21m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/14d5992bbb684c7395efb3491381105f : SUCCESS in 55m 54s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ef194dc0c9a7438790cdffe50be14cb0 : SUCCESS in 44m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f29dbcb7dbbb48eab2a17f83552eedf3 : SUCCESS in 40m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7dd8c0f7058c427ea8c8d9b8631895f0 : SUCCESS in 34m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c1a72da137e4411a636c3d12117fc96 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98740a308ea3439ea3027fa543dfee02 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d01360469844715a0e256a72f221473 : FAILURE in 6m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c31ddac132c4e3eae485557a569df5d : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8e8377677cf7b6e8e48e8ab84d8876fb12d190e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-23 05:33:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0d29d7d45763c4f4935ee1df0a2dc682b4c49c5a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-23 05:34:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nJust fixing up the commit message. LGTM, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"cc4eb0eb312bdc4a4f24cb5f1066281efa00c23b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 06:38:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89ddf5a72034706b22d4a49ac81c2ce : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/08ba6c0d4b5b4437abf52cec295078d9 : SUCCESS in 1h 02m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba83ecea9b184c47bf488b08a2da53ab : SUCCESS in 51m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8792775534ae4eadab1038a9055cf785 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/189a372d8ce244cbb4997b6e83e606c1 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc125acdf7444536bdb5b16b49889f82 : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a5a2d7a35c84c6ba7093687cb7c16c7 : SUCCESS in 13m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce59f2b1db7b4777931d6128f4a302d0 : SUCCESS in 14m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7ad1e336598e41d28e14e0ec7bce7e6d : SUCCESS in 12m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba6a376cf3f041028b5921f81d11b0d6 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88e6051b280d4da59d1911bfa71daa54 : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9c5dfcef6a94bff92d2d7c74b8f821a : SUCCESS in 16m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3d65c26b565c408788111da247435f0a : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/40dcf8d8cd9843799946390aca3a3f76 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2ee5e7ca29c84974a0e34f7dbaaca07d : SUCCESS in 16m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b7858436e3334538ad5a1404d3569b7e : FAILURE in 23m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e567d5d377dc447b9a462e9d354e6dc8 : SUCCESS in 42m 46s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fccd6fa9acc64faaa703691eb441f85d : SUCCESS in 48m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce26f6f336cf40fe867b5948ac0d28c5 : SUCCESS in 37m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c3c3f7cd8994bf39a39b0f2159a58a2 : SUCCESS in 38m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff9249ab8cc14a098aa53c1c45c686a0 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f4f9fb5a8c4705a1e7086b84e83126 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed0150ca615541168fd98f103b00aea3 : FAILURE in 7m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a20d29100fa64deb88be5dc50a29630d : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":3},{"id":"6b6e5892d5cbb27ffd11a630f2cd79dd371a569a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 06:38:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ded0df902a400fbdfff9f6c4da52b75148927dd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 08:05:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065fb1125b1e4455ba5fcc6bc178c8e3 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/31eded0f1a53462fa185c7d871095e7c : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45ddd7e8b91244e98686380cafde931c : SUCCESS in 43m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/85d83eb870214047b2ab28f835b482ea : SUCCESS in 10m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/659adc08b2344e3bb0c175ad3afee499 : SUCCESS in 10m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44be78e553b64f399ac43cb12cc538a6 : SUCCESS in 14m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6943ec1a01a84bb4ae048d777863a382 : SUCCESS in 12m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71159b34e3d54aad98feb3f4ed59d846 : SUCCESS in 14m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a5e74f3e9267475889bc7d37eb979d24 : SUCCESS in 12m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8eab8997bc254690944261fcb289d677 : SUCCESS in 21m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a2bf20c4e5334c3f9752e6e05b48b741 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e135c9150fb64df9b40c7e770b2bacd3 : SUCCESS in 19m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/97283e9bb8634831ac6c498972b79ec2 : SUCCESS in 23m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e979acd7f1e54540beae6be4f121f4c6 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a9ae22487365421fa8b088ff66ca16f3 : SUCCESS in 16m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8fae145e6bfb4974b1418366b8857e8d : SUCCESS in 45m 57s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/83da21a43bd24b27b8224b6ab0f099d2 : FAILURE in 43m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c5992039ee444c496cb348a4395ff0c : RETRY_LIMIT in 20m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e376817a5d994ed7a64e9aafe070e52e : SUCCESS in 43m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da4f2e207dde46f79d7f7ceec5cca9b5 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37203b1a83dd471a831e9abb96f4f7d5 : SUCCESS in 6m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16151b6dd528464aa1018f41a6381b27 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":3},{"id":"38c1731a9770e39338fe95af9e8437e1dd45d097","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-24 05:45:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"76f085af157f62c70dfdaa92d8cb37b616fbab41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 07:15:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce190cb197904dc192e375a5e44a369f : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/58d76b55a0af4736a52a7f89a9ab566c : SUCCESS in 1h 08m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8639fc250a574ad09a2ed7660b67243b : SUCCESS in 58m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/af0dd94fbb114e9faa432b40bbdb173a : SUCCESS in 10m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b575322a1ed241eebe6177a93a92439b : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5b7745bf9194efdb944df24d3ede664 : SUCCESS in 13m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c462ec728d941d78725bf0975f9dd54 : SUCCESS in 12m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1a5a158565ff490b8034c68fa6cbdd8b : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/648fdd219356404f82ee36f2016557c1 : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fef8e22ef4444075a2df28f1105dbe62 : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3ecfbfe232a430e956607e8173413b6 : SUCCESS in 22m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3164ddc97a0f402a9bceb3cea4de9550 : SUCCESS in 20m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bdf62e1834584a1392190ea29e779848 : SUCCESS in 21m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ba4bda116f6c48d29a33bfc8eab5dfc7 : SUCCESS in 20m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3a59658988694aae94686c139e15e5c2 : SUCCESS in 19m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89754bc7d59f4bb19a6a8005b0d25cc3 : RETRY_LIMIT in 4m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5929c06845eb46cfb7060e08a2b0f7cd : RETRY_LIMIT in 5m 05s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/af4d6db743954bab8ce159f8cfa854a0 : SUCCESS in 48m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1cbc74e7091f4a318be98718c52503c5 : SUCCESS in 37m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8412627e6b194319ad4c17823071b405 : SUCCESS in 43m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec1ca3f0f3cf4716983f9c78fb875d92 : RETRY_LIMIT in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b54386e3b9384e69b4e1be94b62e65ad : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ea3fe84c2a24d74962d3db929d2b1de : RETRY_LIMIT in 4m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71c9072073204741a002996003181cb7 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":3},{"id":"940968139d103902c3f09b9dca4bea9b1a9bf913","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-26 17:36:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a030eb7e5fd16007e8f67b4485334eecb605b56e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 22:05:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d369b0c843e94d429851a6368b3daa8d : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/3338f629228243ceb8f09777e3425eed : SUCCESS in 1h 00m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f33d3ba755954e20ad1562dd063d2adb : SUCCESS in 40m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6299160c0764c859f92685c320aa422 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bf8c2082006b4d8e94978197622cf137 : SUCCESS in 13m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a92f8f7353dd42df9bfca729a8ae5a3b : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4549585c5b9b4a9389a90df86ed6ef43 : SUCCESS in 15m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/41f7f6cfafaf4a16a0351c098bed659d : SUCCESS in 20m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fcba95ff369840c483c863a07c85020e : SUCCESS in 15m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1111d4782b6f49fd9633900b18ef9b04 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d2bbad8aa9f4db79f293e8586aa641d : SUCCESS in 23m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a03dda76c5f940cc860f7560199e4a9f : SUCCESS in 18m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/38951d6c2daa448890fe2d42522d29c9 : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fcdfd7ec42fa4a92ac5e9011e2010249 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/097d0d2b04c4496f913abb9315ef6510 : SUCCESS in 17m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3021d1f2d76d498abd29c7c3824200b4 : SUCCESS in 23m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/af8e6cdd10ee4e09b5d563ffc36552d5 : SUCCESS in 49m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2e2d96bbcefb4c33b35e5c8582933f03 : SUCCESS in 44m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f8194f5ca3e472aa42bf085a2179414 : SUCCESS in 1h 04m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e602a09699e4bef958060a13007ba14 : SUCCESS in 40m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1440b89b401747f39cd3e1196dc224ee : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ec080047134919a617599a55a061c0 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b43b8b5a31344e89b1a53e28a239075a : FAILURE in 9m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8085a7c5abb84ab38bc85e0365abc258 : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":3},{"id":"000f479c28ffffefa77c40f5681daf8d2c619845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 22:06:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7370e6d77790f56cd8c4f9e2a29c70088a71e0c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:52:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d40a484b146498e8f31107ee1af0b08 : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/fad9c7bebedf47f699fd7bc624d6c6a3 : SUCCESS in 57m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83b4ba6ba91a44eda6e0c49ea511ecc0 : SUCCESS in 57m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8107ea1d7925474aa68d34774b29ce4f : SUCCESS in 8m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/47efc835d9a54ae29642d7bb63dccf6f : SUCCESS in 12m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1014e2d3878a4afdb68dbdd5c8107661 : SUCCESS in 20m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19f04b98e1d14b5c8c9a550882d1373c : SUCCESS in 21m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/19b300e842474fb2a33540d401da529a : SUCCESS in 20m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e86733838994a7bbb17158ee7c8256a : SUCCESS in 12m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3deef6f92cdf416fbd94efc7db61720a : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/87b6242dc33e4bf9b0ed205e02d274bf : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2cc87772141444189991dd6b34f158ab : SUCCESS in 28m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1c0449e99c8b4adbb511a61a4fa51968 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8ac0c36f562f4830b075b8cb36867ed2 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a592ce16aae24b9598ce87e523b35be3 : SUCCESS in 16m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/266bc81a030b4f0693683f0f25e964d3 : SUCCESS in 49m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b9df60981b5e4664a3b7ee7b01c3cfe6 : SUCCESS in 43m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f26c628ae2c4915abb1a6cb28d80042 : SUCCESS in 35m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35f26e81bffd436ba3bfad8b96e4e94c : SUCCESS in 37m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b8e6c51bfae4903b655c31a8fafd9e8 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfcf70bd8efc441a9f3635c13b0cf935 : SUCCESS in 5m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c27eced9dde94f4dbf939aac8be5ae60 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":3},{"id":"36e0e050f580b49f274c57b2669f00cf6097cf4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:52:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2b0fdfeb6b354a1c34371b35e10cbc6f03f0085b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:53:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa862094433f4c3daa7aba47c9c6fdae : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8f329ca110ad4e32ac022bafa499828e : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/69eb278e2bfa4ebfa48b2963c82e989e : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ea0cab6e3e0e6db8a09b6beea06791d3455a32d1","revisions":{"9f269eabaca32e66ba2aadc72756d141a90be7d7":{"kind":"REWORK","_number":1,"created":"2021-01-21 01:39:26.000000000","uploader":{"_account_id":11534,"name":"david","email":"david.cole@sohonet.com","username":"david-cole"},"ref":"refs/changes/42/771742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/771742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/771742/1"}}},"commit":{"parents":[{"commit":"4ee05c5ddcb214182a1fa3e9e897e1d38c93c21d","subject":"Merge \"s3api: Get rid of slo_enabled flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ee05c5ddcb214182a1fa3e9e897e1d38c93c21d"}]}],"author":{"name":"its-not-a-bug-its-a-feature","email":"david.cole@sohonet.com","date":"2021-01-21 01:22:45.000000000","tz":-480},"committer":{"name":"its-not-a-bug-its-a-feature","email":"david.cole@sohonet.com","date":"2021-01-21 01:39:24.000000000","tz":-480},"subject":"Adjust month comparison from str to int","message":"Adjust month comparison from str to int\n\nswift-drive-audit checks to see if a new year has recently ticked over\nby checking to see if the current month is January and the logs we are\nchecking are in December. datetime.month() does not return a str, it\nreturns an int, so we need to compare with int 1 instead of \u0027Jan\u0027\n\nCloses-Bug: 1912508\nChange-Id: Iabb53f5e4081d580d016bbf75d86e1d75e1f20bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f269eabaca32e66ba2aadc72756d141a90be7d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f269eabaca32e66ba2aadc72756d141a90be7d7"}]},"branch":"refs/heads/master"},"28826e24430b6d78c0cd604b25355e8ceebc12c5":{"kind":"REWORK","_number":2,"created":"2021-01-23 01:15:41.000000000","uploader":{"_account_id":11534,"name":"david","email":"david.cole@sohonet.com","username":"david-cole"},"ref":"refs/changes/42/771742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/771742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/771742/2"}}},"commit":{"parents":[{"commit":"257041affe070c484adb1612c3c5c4eecc4130d7","subject":"Merge \"Fix logging in proxy container GET path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/257041affe070c484adb1612c3c5c4eecc4130d7"}]}],"author":{"name":"its-not-a-bug-its-a-feature","email":"david.cole@sohonet.com","date":"2021-01-21 01:22:45.000000000","tz":-480},"committer":{"name":"its-not-a-bug-its-a-feature","email":"david.cole@sohonet.com","date":"2021-01-23 01:15:16.000000000","tz":-480},"subject":"Adjust month comparison from str to int","message":"Adjust month comparison from str to int\n\nswift-drive-audit checks to see if a new year has recently ticked over\nby checking to see if the current month is January and the logs we are\nchecking are in December. datetime.month() does not return a str, it\nreturns an int, so we need to compare with int 1 instead of \u0027Jan\u0027\n\nCloses-Bug: 1912508\nChange-Id: Iabb53f5e4081d580d016bbf75d86e1d75e1f20bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28826e24430b6d78c0cd604b25355e8ceebc12c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28826e24430b6d78c0cd604b25355e8ceebc12c5"}]},"branch":"refs/heads/master"},"ea0cab6e3e0e6db8a09b6beea06791d3455a32d1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-23 05:33:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/771742/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/771742/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/771742/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/771742/3"}}},"commit":{"parents":[{"commit":"257041affe070c484adb1612c3c5c4eecc4130d7","subject":"Merge \"Fix logging in proxy container GET path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/257041affe070c484adb1612c3c5c4eecc4130d7"}]}],"author":{"name":"its-not-a-bug-its-a-feature","email":"david.cole@sohonet.com","date":"2021-01-21 01:22:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-01-23 05:31:33.000000000","tz":0},"subject":"Adjust initial month value from int to str","message":"Adjust initial month value from int to str\n\nswift-drive-audit checks to see if a new year has recently ticked over\nby checking to see if the current month is January and the logs we are\nchecking are in December. The logs use abbreviated month names, so we\nneed to extract that from \"now\" to make valid comparisons.\n\nCloses-Bug: 1912508\nChange-Id: Iabb53f5e4081d580d016bbf75d86e1d75e1f20bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea0cab6e3e0e6db8a09b6beea06791d3455a32d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea0cab6e3e0e6db8a09b6beea06791d3455a32d1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
