)]}'
{"id":"openstack%2Fswift~771969","triplet_id":"openstack%2Fswift~master~I0ea3c26e54d91f27159805a45e49ad7f8f0e0431","project":"openstack/swift","branch":"master","topic":"p-debug-logger","hashtags":[],"change_id":"I0ea3c26e54d91f27159805a45e49ad7f8f0e0431","subject":"Move DebugLogger to its own module","status":"MERGED","created":"2021-01-22 12:28:52.000000000","updated":"2021-01-27 02:17:56.000000000","submitted":"2021-01-27 02:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"771969-p-debug-logger","meta_rev_id":"44cc4200bb6903ad50342593699b75a6372b5b37","_number":771969,"virtual_id_number":771969,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 02:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-01-24 23:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2021-01-25 15:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2021-01-25 14:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2021-01-25 18:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 13:54:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 15:29:52.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-01-22 20:22:32.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-01-24 23:03:40.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-01-26 17:35:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"3e1e1c0fa9a0bcdbcc27aa28c06dc4dfff960f34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-22 12:28:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c1de73525a9e4ed2f7c288c6b924b9b88e3de28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 13:54:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f8070aa6ff4bca8b8b2c31fc2f79a7 : SUCCESS in 8m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0774307dd590443994e26035e50fdbf9 : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/28c84879eb70482c86b589f4c0744d9e : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a216a4cb31814368897b39a954c99b1d : SUCCESS in 15m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8871f3c83a6c489fb0e621890f8d9c95 : SUCCESS in 16m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c838559794834f8591161c2f57c6cc38 : SUCCESS in 16m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/08e499753f474a8b8173347c7cb99913 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad4951274dda4c1ba86f2b6c801f8833 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf9491f6ca0f46d2bf195ef9bc7414fb : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52531388cc084e509f2a9e2e31d19694 : SUCCESS in 17m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5589c122684847029322f4c26c9fd00e : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3498b76dff054c319eca7cc7cb0cdfdd : SUCCESS in 17m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dabf0c5ae22c404399488ced9cdb3673 : SUCCESS in 18m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3de69dc16f1405c8a3922bbdb25560d : FAILURE in 1h 13m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8b708248dcad4ae5847a52953fb98862 : SUCCESS in 1h 08m 54s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/033576cfd764450797f5acaa38381842 : SUCCESS in 51m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3a0772a388a4d8180fe42bd5dc009fc : SUCCESS in 42m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3de9495b737e433c86f5b94df0ee612e : SUCCESS in 39m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/457ed7b779274e6e9d4e400efda70b28 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3671e55754c42088165e5a536e05136 : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce605ef0a00a40d8b29b9ae8245762af : FAILURE in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0cf259935fb17a470bf03202cfd30a7cb88b7ff3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-22 15:29:52.000000000","message":"Patch Set 1:\n\n(3 comments)\n\ncould maybe rebase https://review.opendev.org/c/openstack/swift/+/771086 on here and drop the mock depends in the playbook","accounts_in_message":[],"_revision_number":1},{"id":"336d5167cf92dbd603b933c3baffc94c1478d7f0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-22 16:58:55.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"026a77fa47de24d381eb366e041b8263f2cf4780","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-22 20:22:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a10fde1c3849844fee86f43f685d168f8eddd9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 22:33:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473d1c613fa0465e914be41930154b47 : SUCCESS in 9m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2feb390670e04184a53c7e0ee775bf67 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6a06d81e4d2647ca85a9940b25d57eed : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc16fc7a907e4627a5a0037ef0b6c996 : SUCCESS in 16m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bc5fdb6db9542168b5761f4b03e4f9e : SUCCESS in 20m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cc4ef987803849eba0587a3f4c91fa15 : SUCCESS in 14m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/626f790f9f0840bb84b73cb86ec5063c : SUCCESS in 12m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e104bdf062e241edb9fef9757a7c6d44 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/06c44de69da0426c81c1e2eb9ef719e9 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/548119d6804c40379c8592ce8453c701 : SUCCESS in 17m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/daaff180b1834ada865e4880348cfe9b : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1b9b5f44c10f4a04a4af8c531c695d46 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4a156e5f58094a38abe50390d9a3a537 : SUCCESS in 20m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b46fa68f1cc43468313b556025d0bab : FAILURE in 25m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/25a2a54b05374a69bd39a374b8a8a51f : SUCCESS in 44m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8d4eb1b0bfc74ddbbd15208ea467d050 : SUCCESS in 41m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acd7deabdcba425e9d0d5831b377bd91 : SUCCESS in 32m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4413d15a7a943d6a9475d745cf4ead6 : SUCCESS in 40m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46020325f53e4940b319036b962c670b : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd8ab802a4764551a6ecf5d3d80e01b3 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f230c44b3e0748bd8f40fb6d13baf8f3 : FAILURE in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd443ac99675a5a777e55b177b4f7b160625e032","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-01-24 23:03:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLove it, nice refactor!","accounts_in_message":[],"_revision_number":2},{"id":"fa2f662faf976a62851578419535e9f0bf17a427","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-25 09:03:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6fa93e62e99689af5679cb98d7bef04ba2b4fcac","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-25 14:49:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nwe should probably land this... the follow up is more debatable:\n\nhttps://review.opendev.org/c/openstack/swift/+/772092","accounts_in_message":[],"_revision_number":2},{"id":"6fde2528f42edb882f9abb31036f21ef3c6dd212","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-25 15:18:32.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7989edb25395545d631d1b04119e39180774d428","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-25 18:28:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"254c39e8f69e5b5e67885b50ee2953e4b0edb97f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 18:29:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d6d5a11baa67015d26a917c11bff88f92e77eaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 01:31:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c4bb68abda44e73ae248b5a064952e9 : SUCCESS in 8m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/46002734fc654f11b25134d3f69fb826 : SUCCESS in 9m 09s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/40dea742f10841f0bb379396ffb15122 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23664a8ae4c44489a73293532507629c : SUCCESS in 14m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d89a579c1684853b2f3e69e07b6cd8f : SUCCESS in 16m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5cac5d1938824bdeb454bf899cd7a17a : SUCCESS in 13m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c084fb5e147e49b98ac836a97d4f23d0 : SUCCESS in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29e57f463e5b4b588988df78ae719859 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0e2441aab97483494e91de6d260c304 : SUCCESS in 22m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b2c222259be1456ca0ff5ad872caad7f : SUCCESS in 19m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/234399d988e54f4486475f0e184a6fdd : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/13f33e3f1c6445ec8543a712c0876072 : SUCCESS in 22m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c52656c4382f48dca1701755660a46d8 : SUCCESS in 17m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/adeac78acb1c41068268d5cea9426e2f : RETRY_LIMIT in 3m 06s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3fdb9e074f4744d5beecb22e60a0b6af : SUCCESS in 51m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f19c5ed80c674eefa7cb862859aa2ee8 : SUCCESS in 41m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1862ded2e21f41d398bc820b4dfe902e : TIMED_OUT in 1h 33m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16be6df66e074843a3324c7f3b228747 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a7ff52f39f4fde84a113d10a06d7d0 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1a3c0f76982aa04e1bcd6e6335c0ea8a12daafac","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-26 17:35:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9c5919911616f392a02ee2bb465022333c63d46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 21:35:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f97a03390da4ea99e0a27bf5eb500a3 : SUCCESS in 8m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf9b7606282b4b739714d02825f11a9c : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9794c3f11b9b47dabead6e05167fd0a9 : SUCCESS in 11m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/350ebd116a8046cf92a0df69650b5493 : SUCCESS in 14m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43c9f65b5ca9457c88823ea3675d5272 : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ef5e735903b4fa58a07baad13ba8ec8 : SUCCESS in 19m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5ce7960a6ba642a9a0f661a50330e45c : SUCCESS in 15m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0f624a902c5e4c4bb64b3e27437f265c : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eac9e806e16e4d04bb77f62e3d2924dd : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/53aa150ff9ee42b784360189cee9ffc2 : SUCCESS in 18m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/74de39bb494c4513b1f763520131dd3f : SUCCESS in 22m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cd8c654211bc4459919af6538445f4ea : SUCCESS in 21m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f3c3ce9c13fb43b9bb6e16d1010dd5db : SUCCESS in 21m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3d1db18a9674d9e8098cf010170f1a6 : SUCCESS in 23m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b797e5d6c65435b8b574aa27b5de2d6 : SUCCESS in 45m 04s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/632cee8175784825a3d736078a2f64f1 : SUCCESS in 41m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8272b54595f44d48ed3f9135ccb56a4 : SUCCESS in 36m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/680b462488bc4bf1b83c7eeacf52a8bb : SUCCESS in 37m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02350c9ebd8c488da64ef34656c9de05 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9adb55615424705aa04559a10f3cdcd : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/356e8f76ade340c389c04ebcfe746164 : FAILURE in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3bf9a0a875d9d439fd0d3d60db23fa09be7146f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 21:35:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"99d258816682e99f1cf24bb5ec10ff0fcc29829b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:15:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f406405e394f467dab295319a1591ff8 : SUCCESS in 8m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c06e104dde7a482ba22c94ba2b815ddf : SUCCESS in 9m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/62d923e75d804610a78177f647117f08 : SUCCESS in 9m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e629148f41144b40800375693421139c : SUCCESS in 12m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56a80962d42f465a9824705611c7f884 : SUCCESS in 14m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/548b4f3d6bbe43ec9a6b5c2371a951d8 : SUCCESS in 13m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1de8143607a24613aeee9f7bf6fcaf06 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd0c5115834c4640bd3457605f42030d : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac1087bbf31e4a9ba721a464857ebfbd : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c59092d37104ca3a43a9cf476b338d5 : SUCCESS in 15m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1328d88e01b140669a50fed0a425acdc : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a15deb22a8964a0d9f93a9e987a4b7a5 : SUCCESS in 22m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6ca28d0a420d46e182f925da8eac7d17 : SUCCESS in 18m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/53989f9b71674ea6a898f581ce0b9c34 : SUCCESS in 57m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0a56104439524a5ea4e3e56754499c03 : SUCCESS in 47m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d11c27c93a9649c19ce8ad533d783d55 : SUCCESS in 47m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a98dd38bcef543f5b500ef40d0d6f0d6 : SUCCESS in 41m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a50df6f5bdf4446cafcedc10ab71209d : SUCCESS in 19m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ca59b703284f239fc00278f05831c7 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a573d8a4ae43a4fa7cf75a82e0fea0f76ae3f5e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:15:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"44cc4200bb6903ad50342593699b75a6372b5b37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:17:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b87ecd5baa314e24a270064ac42bcd71 : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d9e3b680126649cf9fc722a11ef864f9 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/97059d41190f42afb588f31972a96c41 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f4200791beb9a59bd56d41ad4aff5cfce21704d","revisions":{"7038bb0c39ee33370894b176657c8ac84eee0514":{"kind":"REWORK","_number":1,"created":"2021-01-22 12:28:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/69/771969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/771969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/771969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/771969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/771969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/771969/1"}}},"commit":{"parents":[{"commit":"d429918ed08baf0584e34ed122461ce3e433342b","subject":"Merge \"Fix 503s from EC GETs of objects with POST metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d429918ed08baf0584e34ed122461ce3e433342b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-22 12:26:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-22 12:26:24.000000000","tz":0},"subject":"Move DebugLogger to its own module","message":"Move DebugLogger to its own module\n\nMove DebugLogger and associated classes to its own module under test\nso that it can be imported (for example in probe tests) without\nrequiring all the dependencies in test/unit/__init__.py.\n\nChange-Id: I0ea3c26e54d91f27159805a45e49ad7f8f0e0431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7038bb0c39ee33370894b176657c8ac84eee0514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7038bb0c39ee33370894b176657c8ac84eee0514"}]},"branch":"refs/heads/master"},"8f4200791beb9a59bd56d41ad4aff5cfce21704d":{"kind":"REWORK","_number":2,"created":"2021-01-22 20:22:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/69/771969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/771969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/771969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/771969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/771969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/771969/2"}}},"commit":{"parents":[{"commit":"d429918ed08baf0584e34ed122461ce3e433342b","subject":"Merge \"Fix 503s from EC GETs of objects with POST metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d429918ed08baf0584e34ed122461ce3e433342b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-22 12:26:24.000000000","tz":0},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-01-22 16:45:01.000000000","tz":-360},"subject":"Move DebugLogger to its own module","message":"Move DebugLogger to its own module\n\nMove DebugLogger and associated classes to its own module under test\nso that it can be imported (for example in probe tests) without\nrequiring all the dependencies in test/unit/__init__.py.\n\nChange-Id: I0ea3c26e54d91f27159805a45e49ad7f8f0e0431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4200791beb9a59bd56d41ad4aff5cfce21704d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f4200791beb9a59bd56d41ad4aff5cfce21704d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
