)]}'
{"id":"openstack%2Fswift~771975","triplet_id":"openstack%2Fswift~master~Ic08e6e7df7162a4c1283a3ef6e67c3b21a4ce494","project":"openstack/swift","branch":"master","topic":"p-fix-get-from-shard-logs","hashtags":[],"change_id":"Ic08e6e7df7162a4c1283a3ef6e67c3b21a4ce494","subject":"Fix logging in proxy container GET path","status":"MERGED","created":"2021-01-22 13:17:47.000000000","updated":"2021-01-22 17:26:44.000000000","submitted":"2021-01-22 17:08:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"771975-p-fix-get-from-shard-logs","meta_rev_id":"83d5fabc9885675da9b4590355db487aa5d40859","_number":771975,"virtual_id_number":771975,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 17:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-01-22 15:13:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-01-22 15:13:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 14:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 15:13:47.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"32080d04307ff5ead8b0d95838a2e0659d8a471b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-22 13:17:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83bb2c32347f8df28b6a413e91bc0aec819aac9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 14:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3617b1b32b3e4350b80abf47eccccbe1 : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/bd7ea1e3702c41458eaa5463a1e296d0 : SUCCESS in 1h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/beaf5ab8a571431db5302b3864442fb3 : SUCCESS in 50m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/533cec0ce850482ab79ab19b7e202a14 : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c1ec9b5b2b3843b9ac97ac92bed4847b : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dcbb89cf78ca4bda9b8a9738356df0be : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a7af9731dc84a1a9c78facf61becb31 : SUCCESS in 13m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d8fadbaddd34854801d9e941c35074d : SUCCESS in 13m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d3bc88af96c640ae9fe8bd9edee04891 : SUCCESS in 17m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/284b3836b5ad49f1a304e1914389abcb : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03d16a0cb18043f1b359930c060d998f : SUCCESS in 19m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/baec99bd8aaf402bb8788e36e56a3eeb : SUCCESS in 45m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/42dfb59e567243ff812b241a412ad593 : SUCCESS in 20m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2e7987f5fd384e929b9dbf7efe84c75b : SUCCESS in 18m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ea0f76a1f3f148a080a7ead2cd32a518 : SUCCESS in 18m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/541d032e3a9d4a0fa1b90b7aa92a8b67 : SUCCESS in 19m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/83483f5144d04ce4802b2eb79f359647 : SUCCESS in 1h 01m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/cf947328125b4acf9f3489d1a1dbdc1e : SUCCESS in 1h 31m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89ede292f5424214b65dfbd34dabfa23 : SUCCESS in 44m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f729b61e68b943029fd0d0b978a628fb : SUCCESS in 45m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63b1c98fefc64a51b05542cc499708f9 : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ae6125697e48e4bf6a7c4275613889 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48bab8f35a2a4aa98dccca491d56ae1b : FAILURE in 7m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5542e6edf32b4b8c9cc1c4a4ac84f925 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"aca5890deaea69295e51b772d4882c0e329788ef","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-22 15:13:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)\n\nseems like a good solid change to me","accounts_in_message":[],"_revision_number":1},{"id":"1377cbe26c7c9692197504e60d32b0e31182ffb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:14:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78b327893c90feb39cbad1c47a2f02d06f2bb0ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:08:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f27da76ffa4118a9a9c6e0c2cc028c : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/1279381d07e144688b0a9bfd9a5206d7 : SUCCESS in 1h 04m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0558caee01f54ef59b43c259e5f570bb : SUCCESS in 47m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/efb0ddc1386f4e179aa33617b96bd014 : SUCCESS in 13m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/045d3a12991442979a38e322df5d981b : SUCCESS in 12m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/85fa7314c3d64351b691d5fc0bc852cb : SUCCESS in 18m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20a2874d20ee484288162aab40302b5a : SUCCESS in 25m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fc4a890b56054bbf89ea285a6bdda434 : SUCCESS in 13m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a56d7982585d4ed0a97ad15c40018896 : SUCCESS in 18m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3de23f0481a4517be8991a35f38f322 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76f29288cb7c4e6889560787fee64e83 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3f39a8e24184440e8ea0dce7a692057b : SUCCESS in 15m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/df1b449ca0eb4944abac33a0c98d97c7 : SUCCESS in 20m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a36ccb49d86442f1b2a14f922eea04a7 : SUCCESS in 19m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/61ca7852dcd24543ad96c4d77dbfed2b : SUCCESS in 16m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/10c80ad2da60439eb4421a8aabb91215 : SUCCESS in 45m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/35817969c66c4644bc63cc679721dc08 : SUCCESS in 1h 05m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37ab305e9977402681a826fd2036a65a : SUCCESS in 40m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc7cdcf2cb7a44d5ba01d90f1dd38bf5 : SUCCESS in 43m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/427a9d689c4d4aa4a46658cfb9a60cab : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0706d2043a240fe856686c92ef23784 : SUCCESS in 6m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff839e30d44949c9a120fefab9739e4b : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"50177b0519a0a67caae7e93ca54998878069dde3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:08:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83d5fabc9885675da9b4590355db487aa5d40859","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:26:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2178e007a7684ca1b59d42080db4459e : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/64599cf8521b41f0986a136dcd22e769 : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/776a2da8040b4f41ba858a5a0ce1b2c5 : SUCCESS in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a230e5aed3150bcdbf0f9659ee1ab2f74a305eb","revisions":{"0a230e5aed3150bcdbf0f9659ee1ab2f74a305eb":{"kind":"REWORK","_number":1,"created":"2021-01-22 13:17:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/771975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/771975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/771975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/771975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/771975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/771975/1"}}},"commit":{"parents":[{"commit":"d429918ed08baf0584e34ed122461ce3e433342b","subject":"Merge \"Fix 503s from EC GETs of objects with POST metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d429918ed08baf0584e34ed122461ce3e433342b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-22 13:10:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-22 13:10:16.000000000","tz":0},"subject":"Fix logging in proxy container GET path","message":"Fix logging in proxy container GET path\n\nThe proxy tried to map the x-backend-sharding-state header value to\na ShardRange state, but the value is the container *db* state. The\nattempted mapping would commonly fail because db state names do not\nalways correspond to ShardRange state names, in which the case the\nfallback was to log the header value i.e. the correct outcome.\n\nSometimes the attempted mapping would succeed because for example\n\u0027sharded\u0027 is both a db state name and a ShardRange state name. In that\ncase the log message would look something like:\n\n  \"Found 1024 objects in shard (state\u003d(70, \u0027sharded\u0027)), total \u003d 1024\"\n\ni.e. the tuple of ShardRange state name and number was logged, which\nwas inappropriate.\n\nChange-Id: Ic08e6e7df7162a4c1283a3ef6e67c3b21a4ce494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a230e5aed3150bcdbf0f9659ee1ab2f74a305eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a230e5aed3150bcdbf0f9659ee1ab2f74a305eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
