)]}'
{"id":"openstack%2Fswift~772757","triplet_id":"openstack%2Fswift~master~Ie8b923e7be71509e96b80a3873c52609030c73b7","project":"openstack/swift","branch":"master","topic":"cors","hashtags":[],"change_id":"Ie8b923e7be71509e96b80a3873c52609030c73b7","subject":"Add cache-control\u003dno-cache header to CORS test requests","status":"ABANDONED","created":"2021-01-27 18:51:29.000000000","updated":"2021-02-15 09:54:13.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"38c80cce455597b08c835e25cde9f1edb796a5ac","_number":772757,"virtual_id_number":772757,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-03 00:41:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 19:42:37.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-01-27 23:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 22:40:14.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"65a59353d62896ffdb52b6ee5398821e353ef3ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-27 18:51:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23eec816ff5b83df7151f87f64b0a62e3ab91626","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-27 19:42:37.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nAl is the javascript web *master*","accounts_in_message":[],"_revision_number":1},{"id":"79ed649231a02cd2fcd1fecb29b5c1c508d29259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 23:56:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1934008bb24a3f9f232def36ce733c : SUCCESS in 8m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/059214e8b1c945f4811f9aaeb03b77c2 : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/54833fc0aa5f4f6d95facb3a6cc330d9 : SUCCESS in 8m 41s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33f6436993a84acebcbd424fedad65bd : FAILURE in 10m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5da5771a6d534ee0870c0ac70c42c78f : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0463cfba7049403589c67f5d9d3d2043 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9361d9114f03206d2c1bf5f0d9c3933bdedde360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-02 21:36:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9c2211ca9dfa25e8fcfd5917e0b93a1e946cf73d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-02 22:40:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf16225580fdf1cf6d05ccf92c7e091b56376501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 00:41:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9ac83d8988486c81e989a5f94c0c85 : SUCCESS in 8m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6b4c126f12f416dbbac07f335c0b4d6 : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1ccb3906dd31440b93da4ec69589a98d : SUCCESS in 9m 29s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d39e2c0fee84eb79d4a43b997c02e8e : SUCCESS in 7m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c44f56fb49904889a8196778d02c1f61 : SUCCESS in 50m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c13f448c1f64794a9231f1a75603ea1 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"38c80cce455597b08c835e25cde9f1edb796a5ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-15 09:54:13.000000000","message":"Abandoned\n\nused a random query string param here instead https://review.opendev.org/c/openstack/swift/+/533028/22/test/cors/harness.js","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8eea4ae5d9eda68d9e8099160fd677750fc2f2f3","revisions":{"f69cd21e15723398f3c1bb5303f530c5e2ec8e36":{"kind":"REWORK","_number":1,"created":"2021-01-27 18:51:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/772757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/772757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/772757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/772757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/772757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/772757/1"}}},"commit":{"parents":[{"commit":"a19ccbcc705549c3e56370a1d66cc781ebdb88b8","subject":"Add some functional CORS tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a19ccbcc705549c3e56370a1d66cc781ebdb88b8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-27 18:49:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-27 18:49:56.000000000","tz":0},"subject":"Add cache-control\u003dno-cache header to CORS test requests","message":"Add cache-control\u003dno-cache header to CORS test requests\n\nChange-Id: Ie8b923e7be71509e96b80a3873c52609030c73b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f69cd21e15723398f3c1bb5303f530c5e2ec8e36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f69cd21e15723398f3c1bb5303f530c5e2ec8e36"}]},"branch":"refs/heads/master"},"8eea4ae5d9eda68d9e8099160fd677750fc2f2f3":{"kind":"REWORK","_number":2,"created":"2021-02-02 21:36:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/772757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/772757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/772757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/772757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/772757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/772757/2"}}},"commit":{"parents":[{"commit":"a19ccbcc705549c3e56370a1d66cc781ebdb88b8","subject":"Add some functional CORS tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a19ccbcc705549c3e56370a1d66cc781ebdb88b8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-27 18:49:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-02 21:36:24.000000000","tz":0},"subject":"Add cache-control\u003dno-cache header to CORS test requests","message":"Add cache-control\u003dno-cache header to CORS test requests\n\nAlso match expected Etag header to quoted or unquoted values\nto tolerate etag-quoter middleware being present or not.\n\nChange-Id: Ie8b923e7be71509e96b80a3873c52609030c73b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eea4ae5d9eda68d9e8099160fd677750fc2f2f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eea4ae5d9eda68d9e8099160fd677750fc2f2f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
