)]}'
{"id":"openstack%2Fswift~773170","triplet_id":"openstack%2Fswift~master~I014c0c3bfd7ecc204da6e32cb8b6ef60b513c0fd","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I014c0c3bfd7ecc204da6e32cb8b6ef60b513c0fd","subject":"s3api: Flag more request types as known-but-not-supported","status":"NEW","created":"2021-01-29 18:59:05.000000000","updated":"2025-02-07 20:03:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"1c659fa561567cf84b2c3b7865b1a83d251d2add","_number":773170,"virtual_id_number":773170,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-07 19:49:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-29 20:23:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-24 18:16:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"26a8379f2c95839f02fd4c84cd10e5a41a3f57ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-29 18:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0eebb3e62d0094552b9f83e7037eef016b01392d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 20:23:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41dde317523c4f8f986c752629615a71 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/5c0d855a1b304f998be965d133c31d18 : SUCCESS in 1h 05m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4cda16766979495695963cd55134a25c : SUCCESS in 46m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9a485224fdc4265ae67127b43fda8a8 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aabb0730a9d042929e531750528c32c7 : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eca7f54d09d945c6b8addf62a9ca845b : SUCCESS in 14m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8c801be09f04201811eb04dd67d5959 : SUCCESS in 15m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3e12572e2bc345889c8519fa6a41bfaa : SUCCESS in 13m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3521f0522a2e4aacbbeb84a89794f2bb : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/604d9bbf36a548538b9605f8d977572e : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7cb870cbf19d430d9d02de46551500bc : SUCCESS in 20m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6cc78711b9b47d69548288ab10ae591 : SUCCESS in 19m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cb9a40f2de0f40ebb3a5146013d67f07 : SUCCESS in 21m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/968337a37e5048ba89b0602ceba3c5c0 : SUCCESS in 20m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/df4abf82feff4835a53da8a1955fd5ef : SUCCESS in 20m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a4c24d9823904fda938636607f8bbc39 : FAILURE in 23m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2ab97976c194448291143aeca7c088cb : SUCCESS in 45m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9ed74720e6d4412389ea449b4bd3731a : SUCCESS in 50m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9784c5dea934eadb1645236f9e45f9a : SUCCESS in 45m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ad2734f975d48b98382a674da184540 : SUCCESS in 1h 08m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e16cb09fa614432ac2a529360be2518 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/687291846507427a893571b47aad4559 : SUCCESS in 7m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54d4475deb7246399a44fc8fcab2bb54 : FAILURE in 10m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0957ce8fb1ba420489287a0d6e041b45 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ce23888040443a0fde5606e1a5b3ae5c84f7282f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-30 23:49:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66965dd591b5bb2e7af9b8b8cec3f423238d5daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-30 23:49:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d69bcd423fd396f9f19e3bd5c97f28d3469da459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:14:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea4e6d050ab446a85078422ab75e7d9 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/d0630cc3adbc4d5792b79f205cbcd2a8 : SUCCESS in 47m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d2c3c5a46f548f185db8668d4d220cf : SUCCESS in 44m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4868370cba2a4615af77e3c17846a828 : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a2540f58b5c948668751bfd7ffb69886 : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a363988a3855410fa4d4fc1cbc166f62 : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/386f0e9b29ed401394e64fa09628690d : SUCCESS in 21m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b01a606a18ca436180cffb78e6869ea7 : SUCCESS in 25m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6b383d877504a4a956b8b3a9dd7fd09 : SUCCESS in 18m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e795d53a77b5470b8cef37879327c31a : SUCCESS in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/68184fc6bc3642d7b9d315b7c4f95d20 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10cfb0240c6a4f6cbea72fe620308e74 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5e75dca975ac470dab78851fe7fd9c58 : SUCCESS in 17m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d8ff4f0f981f4ee6b5c0aaefd4bcfec8 : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/17e06556a11341ba89dbfa6bf31af356 : SUCCESS in 21m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7995e5a887f748e18a8200b6c5dd5c97 : SUCCESS in 22m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/936ed6230681441f8574b1755d8e7b44 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52ed989cead54039aab8c645971e0e98 : FAILURE in 56m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e57826de1e24e9f95338c16d7067745 : SUCCESS in 50m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f19fb9869f9e4b5aad1a57b3bfeff695 : SUCCESS in 50m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1cc8acf7d6a49129be6cb74f95ec08d : SUCCESS in 45m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/05d7a396ea2c4f7e80a5c60716f81883 : SUCCESS in 45m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeb97b1a87e542c48be69f56e461cadb : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a50764d3914013954f6f84010b827d : SUCCESS in 6m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26d3e421807e44508d6f942fa768ae6a : SUCCESS in 23m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c4027d376404b209e85f6fd19e674b0 : SUCCESS in 45m 51s","accounts_in_message":[],"_revision_number":2},{"id":"aefaadc3e2cadb9fe5aebada8af80627474011ea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-24 18:16:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a83814bd548107f49d53f43fa1a2827670c5f5b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-07 18:40:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"de58b670e67a1240e775a299e6c16f37a267c54d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 19:49:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3573e4f01174074a6f3bdc028f1518d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a4d2d73ff04bf88924b699e94158db : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/d59dac25eab24e788966cbcc1ed1b113 : SUCCESS in 42m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/988204b88ad4434196ba257006c27846 : SUCCESS in 47m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa4e38d8972e49f3b510ecc34e80f4d1 : SUCCESS in 45m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00029c62a8fe406db39ad2c554722cef : SUCCESS in 47m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1673fe334df340c287f38ff2b3682ef2 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d60f30a45b4a4a96a99e9ebc1d05d78c : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ad2f97c9d50482e9d150108c3129c95 : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e197bae5a474cd6bf3e3ba95e8c1316 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2377a0ac85074457a14127a248ba16ea : SUCCESS in 19m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d4c1212da3d4432a85faa98bdfc8494 : SUCCESS in 20m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b1b034a4eb1c4246b626499fed58557c : SUCCESS in 30m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31832cb919b84590b81c9fad33b6a672 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/466ff6179faf409d9c64ca910ae1852c : SUCCESS in 15m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e0e183d21351442c8b4e0e3e3ea9fa02 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b658a4ba60c441e854d9f6c947b82e9 : SUCCESS in 8m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f26a2ec2234c4682a8508b30bee7986e : FAILURE in 17m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/249fef90c34b4cbd92e47725d7241d48 : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/37eba6a6b4a24d02b981ab52c28272d9 : SUCCESS in 52m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f52c0839a5d24a38a3f72c55d059fff1 : SUCCESS in 48m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5133fce3e6c48c29db0cd32f476fba9 : SUCCESS in 48m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fa0c81efc2a488c83e8177f442abe96 : SUCCESS in 26m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd259ac4a24453ca732d96e220d1d5c : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2c2d6d797a94f21aa876069201465d2 : SUCCESS in 21m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6c83eb6633741e8839736cc66bcf417 : SUCCESS in 40m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cef2424480a4705a2a833a5fcaaced9 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1c659fa561567cf84b2c3b7865b1a83d251d2add","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 20:03:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68fed170031448e0a767e9f3c122abbe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7408d3426e51480390faf74fc72583c8 : SUCCESS in 1h 20m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"435f0c1628f9609aba72df9f7c6bd08e298790c9","revisions":{"806c9ec40e8bfe8e8abeb3cf9668cb7e88fe097b":{"kind":"REWORK","_number":1,"created":"2021-01-29 18:59:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/773170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/773170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/773170/1"}}},"commit":{"parents":[{"commit":"833d7f1d1257de852c21f38b226b108739628ef6","subject":"Merge \"Update Memcached Considerations docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/833d7f1d1257de852c21f38b226b108739628ef6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-29 18:58:40.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-29 18:58:52.000000000","tz":-480},"subject":"s3api: Flag more request types as known-but-not-supported","message":"s3api: Flag more request types as known-but-not-supported\n\nWe don\u0027t support them, and the 501 is more obvious than needing to run\ndown why the signatures don\u0027t match.\n\nChange-Id: I014c0c3bfd7ecc204da6e32cb8b6ef60b513c0fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806c9ec40e8bfe8e8abeb3cf9668cb7e88fe097b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806c9ec40e8bfe8e8abeb3cf9668cb7e88fe097b"}]},"branch":"refs/heads/master"},"d6929745ee7d9b99dbf87f69a2783419769788f9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-30 23:49:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/773170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/773170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/773170/2"}}},"commit":{"parents":[{"commit":"833d7f1d1257de852c21f38b226b108739628ef6","subject":"Merge \"Update Memcached Considerations docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/833d7f1d1257de852c21f38b226b108739628ef6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-29 18:58:40.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-03-30 23:49:24.000000000","tz":0},"subject":"s3api: Flag more request types as known-but-not-supported","message":"s3api: Flag more request types as known-but-not-supported\n\nWe don\u0027t support them; we should return a 501.\n\nChange-Id: I014c0c3bfd7ecc204da6e32cb8b6ef60b513c0fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6929745ee7d9b99dbf87f69a2783419769788f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6929745ee7d9b99dbf87f69a2783419769788f9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"435f0c1628f9609aba72df9f7c6bd08e298790c9":{"kind":"REWORK","_number":3,"created":"2025-02-07 18:40:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/773170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/773170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/773170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/773170/3"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-29 18:58:40.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-07 18:40:00.000000000","tz":-480},"subject":"s3api: Flag more request types as known-but-not-supported","message":"s3api: Flag more request types as known-but-not-supported\n\nWe don\u0027t support them; we should return a 501.\n\nChange-Id: I014c0c3bfd7ecc204da6e32cb8b6ef60b513c0fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/435f0c1628f9609aba72df9f7c6bd08e298790c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/435f0c1628f9609aba72df9f7c6bd08e298790c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
