)]}'
{"id":"openstack%2Fswift~773818","triplet_id":"openstack%2Fswift~master~Ib5bf426cf38559091917f2d25f4f60183cd16354","project":"openstack/swift","branch":"master","topic":"relinker","hashtags":[],"change_id":"Ib5bf426cf38559091917f2d25f4f60183cd16354","subject":"relinker: Rehash as we complete partitions","status":"MERGED","created":"2021-02-02 23:21:33.000000000","updated":"2021-02-19 05:01:20.000000000","submitted":"2021-02-19 04:59:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":16,"has_review_started":true,"submission_id":"773818-relinker","meta_rev_id":"d9313e34b85a500cb3c53c078f6cf463aaf6e707","_number":773818,"virtual_id_number":773818,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 04:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-02-19 03:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2021-02-19 03:56:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 01:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 06:04:49.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2021-02-12 15:33:17.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-02-12 16:56:24.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-02-15 06:35:26.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"948f5dbb82cea01bceec5796200d8eb46f1e5cf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-02 23:21:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d1ea0abdc9fdb8854a2ed5f287f24771e4ac1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 01:23:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451bc7a421264395bebebca401de1952 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/18bfb75bee8f40a199c512f8735889a5 : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31cf8a8b8cda4217aa1c648ab444233a : SUCCESS in 44m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cf7405169644a09bbb37e93956c9dcd : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d76ed711722a4441845d4ac7ce4f2618 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bb22f940d9664f338683f6088c0052db : SUCCESS in 13m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe6f260021a04fdf86d29ab1e27bb88a : SUCCESS in 13m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eeeeed69ae744df6b733f4a09cc72d0a : SUCCESS in 15m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0cd4dcccfcc44ecfb9c1518c14a20870 : SUCCESS in 14m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70d40e7024444cb38dd0eb0cd176ec8c : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c683135745794498b49728c6dd3024e1 : SUCCESS in 22m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f1ea66b358b4ca9aad7c26fa5d46afb : SUCCESS in 17m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/00fba0bd05ba49c3935c46d286958b11 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1ec078217cb7407091462c9a4ebee523 : SUCCESS in 20m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4bb4a7b65023424c87fd53a45d768146 : FAILURE in 20m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11652c28c8324bf4bda119806b9a7d37 : SUCCESS in 24m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3a58dcfc4adc49a09e2ecc989077c91a : SUCCESS in 44m 32s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5c44f281ff78471b817fdaade438d65d : SUCCESS in 45m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e66f0cc4712746109b22466dbb75d8e1 : SUCCESS in 49m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3050ba548bde4873803211a55c132b8c : SUCCESS in 36m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61c651c4860d43dfadd9a0de06f11881 : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1806a59016d24a228bda14e840f7a07b : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c843631ce4cf4c3da4ab96d33b0b47b2 : FAILURE in 8m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbeb723ab2304215b82515e1cac9c54b : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f42ee415633c7032b47f07a9e8227c1af5d22b43","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-03 18:55:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1311dbe234fb12d36dd83ccf89897ec7d9284351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 22:18:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/676248470c0141b08507851c0eb55d6b : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/ae901681b76e4a9f91868271ce47d843 : SUCCESS in 1h 06m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edde4c43ee2c4deeb101025ffc15f7a6 : SUCCESS in 43m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/811948e6668c4c3391529bb2a636c921 : SUCCESS in 10m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/42a77b30cad142cda1249d1296d01468 : SUCCESS in 10m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4403acf48a5741d8b3fca2def927d030 : SUCCESS in 12m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3ba9c95e07e48d1be59908e9c338ac4 : SUCCESS in 12m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9b2fcce082d6424abf77027cab30dfa9 : SUCCESS in 14m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/295ccb7b98314741b3c5cd7f02c1d168 : SUCCESS in 12m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fc2d72a21a44faea102fdf57812fed2 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2314fb1fc5ec45a189fd126653c631df : SUCCESS in 27m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4000aa2af1248a18cc15a90569a9f0b : FAILURE in 26m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7b9298a98197476ba98a46ae6b169575 : SUCCESS in 20m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ce6bdc2b2a7b4f219747e4c91dbd2af9 : SUCCESS in 19m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1e185eefdf5249e188f1c48f71f6b157 : SUCCESS in 18m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8d349ebc6ce4ec18bac68528e86eceb : SUCCESS in 22m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/29d11c76f20d42a29f04172ed4ad3c6e : SUCCESS in 49m 45s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d80717db79a441bc83c8d57eb8e6840c : SUCCESS in 48m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d811d6c2a1d422f99929d4ed2047784 : SUCCESS in 55m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/69069bc1719c4784862650f49b089120 : SUCCESS in 56m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e86d064354f485db888e665a7b5d0c0 : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fff5b152d1a64d60ad83c47e920c08a9 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/380a66814919433fa20a29bf60ad3b1b : FAILURE in 11m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b7c4b4d669245e6a1c6962896fd9e8b : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":1},{"id":"718e49d3a7a1406e7c6a44bc3d34cc51e84656e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-04 06:12:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97fe55f49f0c07df772fe2358502a3dff413094a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 07:55:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cbceaedaff84e45861d955a334ba7cb : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/88fc958a3fff469cadb2cb2a0d422d12 : SUCCESS in 59m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e31bea89010343e3bef427a219ce6b4f : SUCCESS in 39m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9279ead4bf0346d5a2d974083f6fb595 : SUCCESS in 11m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/574dab0564cd42208b4c4a6ae732659e : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eb951095cc794e27bdde5cddd1edac46 : SUCCESS in 21m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a4daa5a37bb4349ae6bde2d775e2ad3 : SUCCESS in 18m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/daff9fe8e55e4d29a7f2b275a070747e : SUCCESS in 16m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2461d6787134c34aef58d27cf635637 : SUCCESS in 18m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a772efef168f492eba8f6e1801f1e61e : SUCCESS in 26m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/107b3c3cef0440fdb965305492093f07 : SUCCESS in 24m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa6f749943a34165b2d6517092b4e8a3 : SUCCESS in 20m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1c06c75919d740889ae25a70a95f3cd7 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2068f4f65d0f46e8a3a85552f5273bc2 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/476927bf920f4f7db41433033470688d : SUCCESS in 17m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/938894b36b114f339f7c6ebd1a683237 : SUCCESS in 19m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/32743e0fc15e406a87ee629a6fbb7999 : SUCCESS in 42m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/782b53b66eb9421c9b81e3998eb27790 : SUCCESS in 50m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e734bf35ad6b4a9caff85f349b510f49 : SUCCESS in 44m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa9bc83314ce4aa3924d88b04eda57fd : SUCCESS in 37m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ef920f91057423ba233308c6220e4b2 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c56dac297246918f316795b8de75c3 : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c729c078b1024c30914e6789d0ecb7c3 : FAILURE in 10m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43addc9f5eb84201924a88966138a467 : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8cbaabc2889fd1ba5878ee33d37b8ec3482e8820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-08 23:44:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4f3d1119f66af831f7e24282bd610c3fb28cfc99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:55:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6647966eb8642d496da8e7d1827cfd2 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/70fe481037b440d59509bd55fec08b67 : SUCCESS in 51m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7e8a0e5bfae4c8a91989e66e700947e : SUCCESS in 48m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd671dd8b9e34d37ab0ac3b44207ac1e : SUCCESS in 7m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/859ed76c83c6491aaa6b549750a6a8c9 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a8ef309fba44feb84dfa283fcc6e8c1 : FAILURE in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17fe7e406f804854ab211ee3c87d757d : FAILURE in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c28f85f3dc504ff5a0b96f2dab934671 : FAILURE in 13m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50cc3d90d12c456e9a45b251c7008426 : FAILURE in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c966dde68954f8cae346806d3ef28cf : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c57fefbecba48488c15dd774e4f1373 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7d92272ba3ae4d378ffa8265fb8cf094 : SUCCESS in 16m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1c4f14cb5dc44f3fb0c259f25369061b : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8796ee37db124224b0c050e84c32f78d : SUCCESS in 16m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd10bb8272f74c79a3e89dfe86556fe5 : SUCCESS in 16m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a5d4b3ff1ce48c597c5858fba5606a6 : SUCCESS in 22m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/047beb17cc9f4a30862f0ed8b507f835 : SUCCESS in 46m 29s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e6b6aaa5eade48378126fda6f64b77b4 : SUCCESS in 44m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2284d8b4594e464a9a910e0495f2d4f0 : SUCCESS in 36m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2feb836933e4216a57a5eef2fb304b5 : SUCCESS in 39m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/629d6c369d234f0b80c5b4273a51359a : FAILURE in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35e354474e345f59322af97dadb304d : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c52d2f66f7244cc85d0af005770b4a5 : FAILURE in 8m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a88c4ef1e7124cb4a58100e7178607b9 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b52848956ecb43ed7ba3e3c16e15a15b5b4357ea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-09 17:22:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad7d8842fd4a572a3a8b2bdbe5aaa3651cc0dd6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 21:00:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e27581e1e9455fb68d5ba1d9525092 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/2e04fd250c664e0e9328cecc43959b19 : SUCCESS in 1h 12m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8635b5c9f1aa49d88aa8d57c167030aa : SUCCESS in 50m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3720801460fe47b892892ceb0a0a0166 : SUCCESS in 9m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/065481bfb3114d6c8e084de099bcf0a1 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3124d0ab5ea34dabb21e0e89bbbf2382 : SUCCESS in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d4e181971e845e09882721227773003 : SUCCESS in 14m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ebec059628af499a9444ed947c338649 : SUCCESS in 15m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/09539269596a44119f1f7f1155a3a9b3 : SUCCESS in 15m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5618453ba16249c39fa8dd6dbe763489 : SUCCESS in 24m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/06dfe548f5e74094b0148eff96b50059 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5309316250d241f890ec792a1291ee85 : SUCCESS in 20m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e95a296e170547c7a63dc2d1210d0f1e : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1d4d71a9955048ca997ed5be5b18b0ff : SUCCESS in 18m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/700bfbc969cc451788a3443d9f7925de : SUCCESS in 17m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70b3ee5317a4431fb4ac782b96d3ee2f : SUCCESS in 21m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15f8f008b95744e48aa372369a39d301 : SUCCESS in 47m 02s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8acf5f63742d4df189a832e2d56fc982 : SUCCESS in 49m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4574da9403e94c95900998d525420f0c : SUCCESS in 48m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/877b9d42ecbf46ce8da0f4b153277834 : SUCCESS in 48m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41a9536d727f43f79b4f88d675abd194 : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1e875aa9bc44d1985a43eca58b8d6d : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b092ffbb5d8b4326b67a82087c908d75 : FAILURE in 9m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/404944e5d72b403291347a3612e5de06 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f74b084393599e2214b67194f741010e1347ad7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 00:15:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"166fee6b7a78928d6c4ed58881d56da38216b1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 01:27:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3feea38cf5d34cf1a3487e4a7924b3c4 : SUCCESS in 11m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/bcd4493f38ee4c6a9c5476bffd43e378 : SUCCESS in 1h 08m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/850641c253b24c9b8ee5dabadbf36219 : SUCCESS in 51m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f6f853afda94161b554985af6414439 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/33d1277bbad04ce4be3a8f741e6780e7 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3124779655d148ccbca9bc5817abcbba : SUCCESS in 14m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a09602b3fea141e4b7304c6a9a3ea578 : SUCCESS in 15m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/332cf0500d7b49dd9ce8f1f6652d068d : SUCCESS in 24m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ed9ed13368438a8c51ed7b02ae96c4 : SUCCESS in 14m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1021dde405e4b3899cd8f0a52e05006 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b539bce1106f4b1fa4170c531c9ce949 : FAILURE in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/11ee86ad45da41bb94452e19fefd614e : FAILURE in 29m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/aa82a16f71b64bb286b889ad5c53e31a : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ab5b79d72b814672b307b2902e7f4e3e : FAILURE in 21m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7407c4ff75324615bc5d6ddf93d80cde : SUCCESS in 18m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a283d66904c4a709229189c49eedbcd : FAILURE in 21m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f2c77fc5cac4b47b63f80a951990e3d : SUCCESS in 50m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/88ead3a15fb24bdf82964e7f10af5897 : SUCCESS in 50m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79b692373ed14b95ad8d1d7ca8740a8c : SUCCESS in 33m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46e549cef5684ba380dff0f28e130931 : SUCCESS in 56m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12a70b79b5274fb5a1cf16e2860d0b4b : SUCCESS in 24m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e351a1a7bb4fadae44881a63d6745d : SUCCESS in 7m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/09c345f5895848a5a362223428c9999e : FAILURE in 7m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d2cc83513854134b05626a2e5f9db33 : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a7be187d9fcd8d5795c2c6448665372e532f4be9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 03:10:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"608f29a8fdff00393f50ba454d57ab70c987063c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 04:15:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2686ad5ad6e146db9060ad03c8dab19f : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a5faafc028944e9aad02af73ecbf2ef7 : SUCCESS in 59m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e3099d65ac84f9fae08a19f0c8af349 : SUCCESS in 42m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bd26619c01741af84198da61f9306bb : SUCCESS in 9m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b52801e543f471db169369131804230 : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/938d458dae134fd4a8c700caa788db32 : SUCCESS in 12m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b2871ef718842dc9f9f90b251d18867 : SUCCESS in 11m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f396904242ab439398fc8f70d8f880c3 : SUCCESS in 14m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f1db1216eba340d5bb2bd75ee7a62e61 : SUCCESS in 13m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9c415574abec47a6b2bd82ed89afe157 : SUCCESS in 12m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5bf66851b50546988bfb29180b7163e8 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1d7a7314105e4210a14d84caaa5eb93d : SUCCESS in 22m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96b87a9fcfe64e7d87abefc768ac8e76 : SUCCESS in 14m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/91d38ab974c74fc588afc34717011e9d : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3061ecbab3a34249a2b49b302c72e793 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/91b2c91b30c34fde8f4df429891b4944 : SUCCESS in 15m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e3e1a49cca441d99a5dfb590d4c3c15 : SUCCESS in 19m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/571416820bd743689ffea1cf3d35fdf4 : SUCCESS in 44m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/606d3edaf96447458ab57e794a4e45c9 : SUCCESS in 42m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e4f304202a44ca4a0d79eaa7022dfbe : SUCCESS in 37m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35b18c24b5c84f0d89a3f0403f90a9b8 : SUCCESS in 42m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fe4fec50aa249d09b7d720f45d91358 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea648d9b12f64c4e9b3cc2620f72af65 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76161319be5b423a8b02228a03f8a62e : FAILURE in 9m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3104038e04d479791f8d8cf68be3aa6 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":3},{"id":"21c6e9ec2e317662e9bf34891bad98c7a29c90f7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-11 06:04:49.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nStill need to finish playing with this in my SAIO. but wanted to get my comment up.","accounts_in_message":[],"_revision_number":3},{"id":"ba868998d8e7c1cb37104dc28e84a4b2f0c898b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 19:25:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3fec81111319f1850ecc26d082a5c43f64f0c55f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 19:31:53.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"490373bbf5d2effafc377af9bb59baca959d6040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:07:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4553fa82812d4024b4ff2cf6bf8c12c4 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f9833a96336e45b290c24fc6af968b4d : SUCCESS in 1h 05m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2e718c6900549eb84819a18eb7d6ab8 : SUCCESS in 44m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e0fe949ad3648b59d9834539f8d8dd1 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/60513ea84b704e0aba71eed4df2fcc9e : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa36bde172b14d10ae8de477d744cb1d : SUCCESS in 14m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/932debf8483c4d5298e0daba103707ad : SUCCESS in 14m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/09a99eb8db3c493ebba66b53b085d60e : SUCCESS in 15m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9649ec22d2d04f479cca4798e6f6da44 : SUCCESS in 14m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/94ddf5f0d12e4688beda321839874ecc : SUCCESS in 38m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df739c528f624a4fa2fae424465cd426 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0c4c386007042a68daa5dc315f57c6d : SUCCESS in 16m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88b17099e119419a97f9e1909384ba66 : SUCCESS in 19m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/14c8a67a9e64456b98533f7653fb8025 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e44429014e4b4f5b9b087de437374ee6 : SUCCESS in 26m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/82960fa5a3354158adc266ea3ec89082 : SUCCESS in 20m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9fb51c5f88db4aad84331f0a5b2654c5 : FAILURE in 24m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a20f04585a24ea3aa8bac299a2bca19 : SUCCESS in 51m 03s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/12f833ffdabb48fa9e140970f4699147 : SUCCESS in 50m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81cea3278568404887455443f152c046 : SUCCESS in 41m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82e14689c5da4c52935089b3cfef933f : SUCCESS in 41m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/933b68e92c7a43cc903d259aa2f151d1 : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0601922f0874348b1668c3ee1e52f5a : SUCCESS in 8m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d88a171d8f1408d9b40a9c31d495795 : FAILURE in 10m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/665e55a8ccd948a7b064d31177396114 : SUCCESS in 43m 39s","accounts_in_message":[],"_revision_number":5},{"id":"70719fb951671fe9ac40242a5fd6e09b0664ece4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-12 04:36:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2136c7535c7e5e5110a7e3980a6c0e4d545f27ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-12 15:33:17.000000000","message":"Patch Set 5: Code-Review+1\n\n(5 comments)\n\nLooks good but could do with some test coverage.\n\nTried this out on my SAIO and saw two changes in action:\n\n1. new relinked partition got a hashes.invalid during relink action\n2. new relinked suffix got a hashes.pkl after cleanup action","accounts_in_message":[],"_revision_number":5},{"id":"ba8ac873a00dddc8b434fcb79c7c6b7fc492f119","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-12 16:56:24.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)\n\nI don\u0027t acctually understand how the audit hook code works across stages to rehash everyone; but perhaps some tests would make it more obvious to me","accounts_in_message":[],"_revision_number":5},{"id":"a7aed7c5fa8a0297f5818c9cf9e83bf243801891","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-12 17:41:28.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)\n\nTim explained more over voice - we just need more tests","accounts_in_message":[],"_revision_number":5},{"id":"23679b6de1967d178c05ff115fa73cbde9779780","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 22:00:49.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c91ca7bac4d23ce3ed1ab0ac380b2b64e0dead97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 23:50:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9843d282b2c77951fc3754de6a722a3d8eb37abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 00:52:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1613bb27be024887bd0f41a077543275 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8fbcb1f0f6ca4de7b813fa3d40103f5d : SUCCESS in 1h 00m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a2598c3b0e6646c0be4aa4695985d029 : SUCCESS in 39m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5045dbe2f0dc4555a858572fcaf28e69 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2309585df2a04f09838cd31ab34476b9 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf2ee651cd4341e287447b99024fd4b2 : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6cbd40e704746e5a14bb1a1dc7c4240 : SUCCESS in 14m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/54488014fb27443fbd00a28af1d3a7eb : SUCCESS in 14m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dcc49a2a20bd4421afa63b340f3e49b8 : SUCCESS in 13m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2925bbcd8f844b06899c561567cf5dcb : SUCCESS in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3bf84eb3c3224c099cb70a0a7418332a : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/750d1063043f40459e89b049c699a17b : SUCCESS in 15m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e96d3b4aeebe4851be3593c6f78e2dfa : FAILURE in 19m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7e730d0d5b5741828a7c93f2e0692e30 : SUCCESS in 22m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8b46598c5c724691b1a85011b708c902 : SUCCESS in 22m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6d12ae0332df45079fba6d46d4f1e3ad : SUCCESS in 17m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69af7059e2234d27befa6dcee6798b96 : SUCCESS in 19m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b339fd111d74641beb925b08aeef377 : SUCCESS in 45m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6a12684946954258b6550f26150eccf5 : SUCCESS in 45m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d16e401afa3e468abad8645878f8010b : SUCCESS in 35m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d466ae5eb7a14b3e9ae08bdfd12264d9 : SUCCESS in 45m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f3395fc98af4726861e8a7e30a65c42 : SUCCESS in 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beaa26dc526c4f6abb8fb0b78aee14ff : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c2131ad1dd41438b996a4edbf9a95909 : FAILURE in 8m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a6963707b4845e6b08ad4da1ffbbe79 : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":6},{"id":"09051761585e1f3f0c0fdff85be934e4352dd28b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-13 03:14:56.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"20133c7cc9535548c47987274acf58e79e1a1ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 04:29:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e27817d197a43b69bbbc4f36dab0799 : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/66cb2adec0e14249a23fa961f631b5b5 : SUCCESS in 1h 09m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c4f68109f0b3401fad82727ea4a9c520 : SUCCESS in 54m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0fccc03a2cc489482d9c45fbaf3a47c : SUCCESS in 7m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/044dec6e764746598df75df486cd422a : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b5f8dac178f4e82a7f5a796659a0f8b : SUCCESS in 13m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ca195e7adcf425ba04ab26afc4eeb94 : SUCCESS in 14m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68ce60fe341a4528b045540b2cabcec7 : SUCCESS in 13m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8a3e338f0938425b8cb73c93ec416096 : SUCCESS in 17m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57142788bfdc4793828d79607f583119 : SUCCESS in 14m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de001070d0064ce19d8d623b66ec5c43 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1350874af6364145b5d5039450b1d939 : SUCCESS in 20m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2de2fa7a1684d22a894d9ee81c202fc : SUCCESS in 16m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2537b369b501469eb830be41b04ec0a1 : SUCCESS in 25m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6793ddbf5e524d908ee8fc8a3f6dd1d8 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d85a68b89bb047a8beec1f0bb95ec5da : SUCCESS in 18m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/60a72b91bbb14d6fac85aceeb73f5de1 : FAILURE in 21m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0123b1417ef74d359bfe2cb2d4fa8d05 : SUCCESS in 42m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b7d0faf228fb4d79950af4af7f4ac33e : SUCCESS in 45m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bf97393c4804e6fbae0dd80209f7cdf : SUCCESS in 34m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5fceab5a1cc4821b186bd6825844dc3 : SUCCESS in 40m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d32eeff41e24d57b9294b8267cc0988 : SUCCESS in 23m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31532a66ba1244bca08291f2a144c8a1 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8f8604683404f6b86758f5712e13e1a : FAILURE in 7m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5425fe4d6a5c40b4af8b5c176e66f80c : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":6},{"id":"51ebe9c4307c61463355bc6ec7f0df442a7ad75b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-13 05:29:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d600d06aa158d35740bd3712f21d0fba25950d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 06:37:26.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e0d2e3d254746c68d84dd1ffcc7a43f : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/eb78768c3fc34f44a324ddb1af489e5a : SUCCESS in 57m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7db645d1e914afd85eaeb0847f6a95b : SUCCESS in 37m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19360adc2b34467481aa2cc80b3e7c2b : SUCCESS in 7m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e0b6bedd300141f1bec0cac59c7bdd6b : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3d57076cfba44a73a6a7cf725f648586 : SUCCESS in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00188c689ca3441bac108eb7efc6fe16 : SUCCESS in 27m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fb947efe4770490bbae97cdf02b183a3 : SUCCESS in 29m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/63f3f0fefe66490785250ece452f4736 : SUCCESS in 29m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eff28b9fca3b4027a5c1390723f1b161 : SUCCESS in 30m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/649d949713ca44808a9bc35dbdfb5108 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b65eba7bf48748e8b73e6a890009fdae : SUCCESS in 31m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ebcfc97798e84c52b94068592b79959b : SUCCESS in 19m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/243dd2f52e784b17b37257b4660354ae : SUCCESS in 29m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/60116a9bcace4c139f50ebb210a4b4a4 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b55f5b829642412eb3c8bfda07be5e4c : SUCCESS in 16m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61c0ad13ffc540fdbb66a873ed8fab96 : FAILURE in 34m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5251fe9e43704235b424bbe92210cad3 : SUCCESS in 47m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3726d4be6be042bebcf490682ba1bbd8 : SUCCESS in 1h 02m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94ad747bcb724b879e025fd558c34445 : SUCCESS in 43m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99bc9fe412294d259d586d57cf128bcd : SUCCESS in 32m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0400c9abb0214bc1879f28e790eb3a5d : SUCCESS in 14m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5952e89fb834762bf24a2201212e570 : FAILURE in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d1231c320cf44e2b9a6c6246bde4f17 : FAILURE in 7m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f885449bd7b4e8eb197305a6d57d409 : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":7},{"id":"4524997cc32b055acbf2f53961da7e2895241661","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-15 06:35:26.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nNeed to go make some dinner for the kids, so will play with this more tomorrow in actaul testing rather then just loading relinking and hashing in my head.","accounts_in_message":[],"_revision_number":7},{"id":"77c888d3491fec333757e425dc786885b91e6339","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-15 06:57:01.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"46cd065646635a1a92bb8973367946f701ea23a0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-15 10:05:15.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)\n\noh, of course 2**part-power -1 would be 1/2 of the lower... and after reading it again it all just clicked! very clever! I see it now, you shift into the upper space an rehash. that\u0027s a genius move.\n\nJust want to test it properly in my SAIO and not going to finish doing that now (it\u0027s getting late here)","accounts_in_message":[],"_revision_number":7},{"id":"3c88fe860e436a973c262fb68fdf5b7f58d5a240","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-16 07:35:24.000000000","message":"Patch Set 7: Code-Review+2\n\nI was a little worried that it could somehow rehash before all new objects have been placed there. but because in the upper 1/2 everything splits evenly 1 -\u003e 2 partitions it\u0027s fine.\n\nNice one Tim.","accounts_in_message":[],"_revision_number":7},{"id":"0355d72a3e4446880c90298712f4d1e4bd5489c6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-18 17:44:37.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b5d4f8b81dfe3fb73a203f73e0a421ccb533f0a8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-18 17:46:09.000000000","message":"Patch Set 7:\n\noops, meant to comment\n\nI like the additional tests here https://review.opendev.org/c/openstack/swift/+/775504/2 although I now have a hand i them so would say that :) so maybe they could be squashed in. That patch also fixes the pep8 errors.","accounts_in_message":[],"_revision_number":7},{"id":"c107b1dd55781735994e4b5c89ed52c2296660f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-18 19:48:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"09deee2378647864f85eeebf5782e406ce28b078","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-18 19:54:57.000000000","message":"Patch Set 8:\n\nfixed the pep8 errors so dependent patches can pass tests","accounts_in_message":[],"_revision_number":8},{"id":"f4b027d6c49e975bd480a1ef639cacddd7b0be4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 00:59:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9cf7b4b87a4e9895bb3fba3bd2c9f6 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/43630763b375450ba55173dd40df28b6 : SUCCESS in 1h 09m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/061590447bc941dba82f1d971a7cb482 : SUCCESS in 41m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/95f4382ee4b94f828a517f4c263e4354 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/071156694aa746f3abed30d60d6c3e97 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47ff6201bbb24961acfa9a2e3bf2a7b6 : SUCCESS in 14m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ba074f9e3c470797d04f49fb65d4f8 : SUCCESS in 17m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b634520086b44c5ea13128e3861dcf49 : SUCCESS in 18m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4dbe22b6fb5847fca24cea4d8f10fd45 : SUCCESS in 15m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/30d3464c60954684b0c02aa7d2d8c1a5 : SUCCESS in 17m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ff91414942824288ac1de0210a904a04 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f5f5bb6608a48969fd1f98e25c13058 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1125a1056041447b951920846114670f : SUCCESS in 18m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/03006fc6e74f483eb6c29fac1096b29a : SUCCESS in 24m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1de0ad0f5aae42c1b531799f05f1d917 : SUCCESS in 19m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d8c5496073fa4abcb5bf69bd3848eb7f : SUCCESS in 19m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ced78afd34aa432d8ac0d4ae62da7587 : FAILURE in 23m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a76e14513dbf4a12bc0a18dc5eadbf5d : SUCCESS in 52m 54s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e8d6252c6d7c4927abcfbc71394e545a : SUCCESS in 48m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/458123722c404c3bbff011d8043f7e5d : SUCCESS in 35m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de15d3cb14084873bb4bdd939f102f63 : SUCCESS in 45m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64b39d9a89cd40aeba3bf4873e462c4f : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb953db0019c4f4b838e616c3a817b0d : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa4726ea9270459c8946e426ac53fc80 : FAILURE in 11m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad459725a8f04412abf3121d7bfea535 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":8},{"id":"6ee5ed0c076a5dcc6d8dfe0ba30eb7bab8ea2be5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-19 03:47:43.000000000","message":"Patch Set 8: Code-Review+2\n\nLooks great, still got my +2, does what it says on the tin.","accounts_in_message":[],"_revision_number":8},{"id":"0a1e7492996e0db1c8663b5e8e19109323728d3a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-19 03:56:04.000000000","message":"Patch Set 8: Workflow+1\n\nI\u0027m going to make an executive decision and let\u0027s finish landing this chain. There are now patches on top already +Aed. So let\u0027s get a\u0027mergin.","accounts_in_message":[],"_revision_number":8},{"id":"b71ed158edd72d1259bc8027d01ceaaaefa49836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:56:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"77b709baa6e7e5c215086cd300d81b3ed1455c57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:59:42.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb5a0f9641e4d8ea7640ad07408612e : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/c86f2274b2ff40b5b4768d4c1f1ac986 : SUCCESS in 1h 01m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/417e829b21344e1ead426a3978321d0d : SUCCESS in 40m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/15a584ce1bd94d228617c4bbe8aa9b47 : SUCCESS in 8m 33s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/60d3eccbb7f94c6fadddb87f84a90c9f : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9599f4dd229d486395c32c4c47d239d0 : SUCCESS in 15m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e9e47a1e7f0f41108a139e87a18147d5 : SUCCESS in 18m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c164a3c236f641cf988468c669f004ca : SUCCESS in 23m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/002cc69ee789403884cb9c504e33fe0b : SUCCESS in 14m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c012621064ae4b559f185ef10785408f : SUCCESS in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1101f6bc96814462a325d16425dbd7bd : SUCCESS in 18m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/23942a4876d5421680ebb86866b32c3c : SUCCESS in 30m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b48e9b48033e441f96a10eb5be2e13ec : SUCCESS in 21m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/234a12c66fa644ad996d8f3ddb1159aa : SUCCESS in 30m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d9951a060a9d4811a086469914cf7c09 : SUCCESS in 20m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f079c50f29e54c61bb1bf914779af0e2 : SUCCESS in 26m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cb45cbd9f9ff4b9db89c8f1b67819147 : SUCCESS in 56m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8bbd913e2f3b4025a37b7c3ef57b5ae1 : SUCCESS in 55m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86ea595d7b994a87b247808e5a25dd02 : SUCCESS in 32m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/958efeacc09f4be5a0a64567c0df7940 : SUCCESS in 37m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/604d49110a0e4cf893d080c06e7023ea : SUCCESS in 20m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300d38e1b9394696b58e5d969a4ae7e1 : SUCCESS in 5m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0594e106054e45048219d7c87f816990 : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":8},{"id":"7e76c1306a2725a86785f4288757e5dd88bf44a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:59:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"d9313e34b85a500cb3c53c078f6cf463aaf6e707","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 05:01:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf3ebf123f6463ab48dd5772d0fa3d4 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2591a5442f2540688049fd664031eba8 : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8db3feabb8424ae79366be1670b8eccf : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d24884450f6e852a6d9626e45892ca9208783cb9","revisions":{"e36401744af1dceccede54d7408241833908063b":{"kind":"REWORK","_number":1,"created":"2021-02-02 23:21:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/773818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/1"}}},"commit":{"parents":[{"commit":"7306514e04565725dfd3f773e9c0a60c023baa9d","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7306514e04565725dfd3f773e9c0a60c023baa9d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:19:06.000000000","tz":-480},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--objects-per-second.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e36401744af1dceccede54d7408241833908063b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e36401744af1dceccede54d7408241833908063b"}]},"branch":"refs/heads/master"},"92d9f8a87c9de647cc7a91fea63f9ac82fe329cd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-08 23:44:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/773818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/2"}}},"commit":{"parents":[{"commit":"ae721ce5c774d840183a9ca4f9166d5bb105d666","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae721ce5c774d840183a9ca4f9166d5bb105d666"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-08 23:43:15.000000000","tz":-480},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--objects-per-second.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d9f8a87c9de647cc7a91fea63f9ac82fe329cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d9f8a87c9de647cc7a91fea63f9ac82fe329cd"}]},"branch":"refs/heads/master"},"2fa1f68d6aec1524debee20241b06a25fa612081":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-11 00:15:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/773818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/3"}}},"commit":{"parents":[{"commit":"16a45186c1a43d862606fbf7fd1471b8762e327f","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16a45186c1a43d862606fbf7fd1471b8762e327f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-11 00:14:33.000000000","tz":-480},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--objects-per-second.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fa1f68d6aec1524debee20241b06a25fa612081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fa1f68d6aec1524debee20241b06a25fa612081"}]},"branch":"refs/heads/master"},"47c97e8ee8facbffcf47d8ec5c562f166b70cb1f":{"kind":"REWORK","_number":4,"created":"2021-02-11 19:25:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/773818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/4"}}},"commit":{"parents":[{"commit":"5ca1157ec65c6690a43efb7b8ab892e400761b43","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca1157ec65c6690a43efb7b8ab892e400761b43"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-11 19:24:51.000000000","tz":-480},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--files-per-second.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c97e8ee8facbffcf47d8ec5c562f166b70cb1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c97e8ee8facbffcf47d8ec5c562f166b70cb1f"}]},"branch":"refs/heads/master"},"bd856bc63c1cc09324092742a3a5ad5bca1bf41a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-11 19:31:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/773818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/5"}}},"commit":{"parents":[{"commit":"a90515bbbe2db6914229e03da8937e35114e0adb","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a90515bbbe2db6914229e03da8937e35114e0adb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-11 19:31:39.000000000","tz":-480},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--files-per-second.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd856bc63c1cc09324092742a3a5ad5bca1bf41a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd856bc63c1cc09324092742a3a5ad5bca1bf41a"}]},"branch":"refs/heads/master"},"5c8608caef16fcc30b06ad12bb591e1132d58ba0":{"kind":"REWORK","_number":6,"created":"2021-02-12 23:50:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/773818/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/6"}}},"commit":{"parents":[{"commit":"a90515bbbe2db6914229e03da8937e35114e0adb","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a90515bbbe2db6914229e03da8937e35114e0adb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-12 23:21:53.000000000","tz":-480},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--files-per-second.\n\nAlso clean up empty partitions as we clean up. Operators with metrics on\ncluster-wide primary/handoff partition counts can use them to monitor\nrelinking/cleanup progress:\n\n P  3N            .----------.\n a               /            \u0027.\n r              /               \u0027.\n t  2N         /       .-----------------\n i            /        |\n t           /         |\n i   N -----\u0027---------\u0027\n o\n n\n s   0 ----------------------------------\n       t0   t1    t2   t3   t4    t5\n\nAt t0, prior to relinking, there are\n\n    N :\u003d \u003creplica count\u003e * 2 ** \u003cold part power\u003e\n\nprimary partitions throughout the cluster and a negligible number of\nhandoffs.\n\nAt t1, relinking begins. In any non-trivial, low-replica-count cluster,\nthe probability of a new-part-power partition being assigned to the same\ndevice as its old-part-power equivalent is low, so handoffs grow while\nprimaries remain constant.\n\nAt t2, relinking is complete. The total number of partitions is now 3N\n(N primaries + 2N handoffs).\n\nAt t3, the ring with increased part power is distributed. The notion of\nwhat\u0027s a handoff and what\u0027s a primary inverts.\n\nAt t4, cleanup begins. The \"handoffs\" are cleaned up as hard links and\nnow-empty partitions are removed.\n\nAt t5, cleanup is complete and there are now 2N total partitions.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c8608caef16fcc30b06ad12bb591e1132d58ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c8608caef16fcc30b06ad12bb591e1132d58ba0"}]},"branch":"refs/heads/master"},"178c5ae456bc7056c6660008cf6d9e100343ef9c":{"kind":"REWORK","_number":7,"created":"2021-02-13 05:29:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/773818/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/7"}}},"commit":{"parents":[{"commit":"a90515bbbe2db6914229e03da8937e35114e0adb","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a90515bbbe2db6914229e03da8937e35114e0adb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-13 05:24:09.000000000","tz":-480},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--files-per-second.\n\nAlso clean up empty partitions as we clean up. Operators with metrics on\ncluster-wide primary/handoff partition counts can use them to monitor\nrelinking/cleanup progress:\n\n P  3N            .----------.\n a               /            \u0027.\n r              /               \u0027.\n t  2N         /       .-----------------\n i            /        |\n t           /         |\n i   N -----\u0027---------\u0027\n o\n n\n s   0 ----------------------------------\n       t0   t1    t2   t3   t4    t5\n\nAt t0, prior to relinking, there are\n\n    N :\u003d \u003creplica count\u003e * 2 ** \u003cold part power\u003e\n\nprimary partitions throughout the cluster and a negligible number of\nhandoffs.\n\nAt t1, relinking begins. In any non-trivial, low-replica-count cluster,\nthe probability of a new-part-power partition being assigned to the same\ndevice as its old-part-power equivalent is low, so handoffs grow while\nprimaries remain constant.\n\nAt t2, relinking is complete. The total number of partitions is now 3N\n(N primaries + 2N handoffs).\n\nAt t3, the ring with increased part power is distributed. The notion of\nwhat\u0027s a handoff and what\u0027s a primary inverts.\n\nAt t4, cleanup begins. The \"handoffs\" are cleaned up as hard links and\nnow-empty partitions are removed.\n\nAt t5, cleanup is complete and there are now 2N total partitions.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/178c5ae456bc7056c6660008cf6d9e100343ef9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/178c5ae456bc7056c6660008cf6d9e100343ef9c"}]},"branch":"refs/heads/master"},"d24884450f6e852a6d9626e45892ca9208783cb9":{"kind":"REWORK","_number":8,"created":"2021-02-18 19:48:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/773818/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/773818/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/773818/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/773818/8"}}},"commit":{"parents":[{"commit":"a90515bbbe2db6914229e03da8937e35114e0adb","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a90515bbbe2db6914229e03da8937e35114e0adb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-02 23:16:43.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-18 19:46:57.000000000","tz":0},"subject":"relinker: Rehash as we complete partitions","message":"relinker: Rehash as we complete partitions\n\nPreviously, we would rely on replication/reconstruction to rehash\nonce the part power increase was complete. This would lead to large\nI/O spikes if operators didn\u0027t proactively tune down replication.\n\nNow, do the rehashing in the relinker as it completes work. Operators\nshould already be mindful of the relinker\u0027s I/O usage and are likely\nlimiting it through some combination of cgroups, ionice, and/or\n--files-per-second.\n\nAlso clean up empty partitions as we clean up. Operators with metrics on\ncluster-wide primary/handoff partition counts can use them to monitor\nrelinking/cleanup progress:\n\n P  3N            .----------.\n a               /            \u0027.\n r              /               \u0027.\n t  2N         /       .-----------------\n i            /        |\n t           /         |\n i   N -----\u0027---------\u0027\n o\n n\n s   0 ----------------------------------\n       t0   t1    t2   t3   t4    t5\n\nAt t0, prior to relinking, there are\n\n    N :\u003d \u003creplica count\u003e * 2 ** \u003cold part power\u003e\n\nprimary partitions throughout the cluster and a negligible number of\nhandoffs.\n\nAt t1, relinking begins. In any non-trivial, low-replica-count cluster,\nthe probability of a new-part-power partition being assigned to the same\ndevice as its old-part-power equivalent is low, so handoffs grow while\nprimaries remain constant.\n\nAt t2, relinking is complete. The total number of partitions is now 3N\n(N primaries + 2N handoffs).\n\nAt t3, the ring with increased part power is distributed. The notion of\nwhat\u0027s a handoff and what\u0027s a primary inverts.\n\nAt t4, cleanup begins. The \"handoffs\" are cleaned up as hard links and\nnow-empty partitions are removed.\n\nAt t5, cleanup is complete and there are now 2N total partitions.\n\nChange-Id: Ib5bf426cf38559091917f2d25f4f60183cd16354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24884450f6e852a6d9626e45892ca9208783cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24884450f6e852a6d9626e45892ca9208783cb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
