)]}'
{"id":"openstack%2Fswift~773934","triplet_id":"openstack%2Fswift~master~I74cbb8bed2d1fe46cb2a7d7fd851bb4454518045","project":"openstack/swift","branch":"master","topic":"p-cors-test-random-query-string","hashtags":[],"change_id":"I74cbb8bed2d1fe46cb2a7d7fd851bb4454518045","subject":"Add random query string to CORS test requests","status":"ABANDONED","created":"2021-02-03 15:59:59.000000000","updated":"2021-02-05 19:04:09.000000000","total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"0526b82802c48b414854c3a75f67d31eb47b599c","_number":773934,"virtual_id_number":773934,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-03 21:17:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-02-03 18:41:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-03 18:35:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 18:35:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-03 18:41:41.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-02-03 21:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb2bea8bd0d9e0bf7802a0380eef3f753ed4ee44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-03 15:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bfc12670cec18b7eaa7a00e0dc7b927b5a87ad6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-03 18:35:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nSeems quite reasonable. Just a couple questions (honestly; I\u0027m not saying anything needs to change ;-)","accounts_in_message":[],"_revision_number":1},{"id":"aa96f9560e449a59a025d64669a0d65bebf20677","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-03 18:41:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nI. Have.  NO IDEA.  What I\u0027m doing.  WEEEeeee","accounts_in_message":[],"_revision_number":1},{"id":"af4707bdcb65578cd9a419efdda560d94f881e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 21:17:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45a1e650f514be8b9bbe743231903ea : SUCCESS in 9m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52c121d225ea46208e0bf5b222bcace9 : SUCCESS in 11m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e018bcef545f47b2a102d51abafd07b5 : SUCCESS in 9m 11s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5e11344898d4c1886932cc9bb144df5 : SUCCESS in 15m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/847f5c6b60cc4ca2bb31608a55a632e7 : SUCCESS in 1h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/855510263eca45b19d73765adc9740c1 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0526b82802c48b414854c3a75f67d31eb47b599c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-05 19:04:09.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/533028","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04574c8fe4ac3ebc0415c482b90dcad4d2ba9279","revisions":{"04574c8fe4ac3ebc0415c482b90dcad4d2ba9279":{"kind":"REWORK","_number":1,"created":"2021-02-03 15:59:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/773934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/773934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/773934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/773934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/773934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/773934/1"}}},"commit":{"parents":[{"commit":"a19ccbcc705549c3e56370a1d66cc781ebdb88b8","subject":"Add some functional CORS tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a19ccbcc705549c3e56370a1d66cc781ebdb88b8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-03 15:45:58.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-03 15:45:58.000000000","tz":0},"subject":"Add random query string to CORS test requests","message":"Add random query string to CORS test requests\n\nAdd random and date params to CORS test query strings\nto avoid the request being served from browser cache.\n\nChange-Id: I74cbb8bed2d1fe46cb2a7d7fd851bb4454518045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04574c8fe4ac3ebc0415c482b90dcad4d2ba9279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04574c8fe4ac3ebc0415c482b90dcad4d2ba9279"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
