)]}'
{"id":"openstack%2Fswift~773935","triplet_id":"openstack%2Fswift~master~I191307967813ed77d41d12900aa641a2d7735069","project":"openstack/swift","branch":"master","topic":"p-cors-test-random-query-string","hashtags":[],"change_id":"I191307967813ed77d41d12900aa641a2d7735069","subject":"Make CORS tests tolerate quoted or unquoted Etag","status":"ABANDONED","created":"2021-02-03 16:05:05.000000000","updated":"2021-02-05 19:04:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1fcacfe9141b85f4153f1db6647afb157f8f0344","_number":773935,"virtual_id_number":773935,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-03 21:08:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-03 18:27:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-03 18:27:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 18:27:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-03 21:08:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4f3261b20d3206db0dc80e9247c249eb3340284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-03 16:05:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c67d3de05b60d0d6fde2d98fc3e5dbe180f0ec54","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-03 18:27:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGood call; I\u0027m pretty sure I wrote these before etag-quoter was a thing.","accounts_in_message":[],"_revision_number":1},{"id":"a18bccc73b94e6a0fbbf2a76d05fb919ca773567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 21:08:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d91929d93544cd786aeb179f56775e8 : SUCCESS in 8m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a5d8dd20f1b4bb0916fe87c014b3cd5 : SUCCESS in 11m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/03882a4d8e4247c39dc2264f55a1f2ff : SUCCESS in 10m 16s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7775593b768436e83a05e3f594d4245 : SUCCESS in 13m 06s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/99c80f7da115455cae0c811187c33000 : SUCCESS in 50m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d4b8c2193ef46f7ab4873c9522a321a : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1fcacfe9141b85f4153f1db6647afb157f8f0344","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-05 19:04:18.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/533028","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53e40785bcb32b75f5d882d993005cc663e75c94","revisions":{"53e40785bcb32b75f5d882d993005cc663e75c94":{"kind":"REWORK","_number":1,"created":"2021-02-03 16:05:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/773935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/773935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/773935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/773935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/773935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/773935/1"}}},"commit":{"parents":[{"commit":"04574c8fe4ac3ebc0415c482b90dcad4d2ba9279","subject":"Add random query string to CORS test requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04574c8fe4ac3ebc0415c482b90dcad4d2ba9279"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-03 16:03:01.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-03 16:03:01.000000000","tz":0},"subject":"Make CORS tests tolerate quoted or unquoted Etag","message":"Make CORS tests tolerate quoted or unquoted Etag\n\nMake the CORS tests match expected Etag against quoted or unquoted\nresponse values in order to tolerate the etag-quoter being in the\nproxy pipeline or not.\n\nChange-Id: I191307967813ed77d41d12900aa641a2d7735069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53e40785bcb32b75f5d882d993005cc663e75c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53e40785bcb32b75f5d882d993005cc663e75c94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
