)]}'
{"id":"openstack%2Fswift~773982","triplet_id":"openstack%2Fswift~master~I17ef7c2d1a71479b7dca68a3a5c83b5a0e296e09","project":"openstack/swift","branch":"master","topic":"p-cors-test-random-query-string","hashtags":[],"change_id":"I17ef7c2d1a71479b7dca68a3a5c83b5a0e296e09","subject":"use some javascript builtins","status":"ABANDONED","created":"2021-02-03 18:40:39.000000000","updated":"2021-02-05 19:04:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84916996fc22dfe6bacd1ac74d2750481218d128","_number":773982,"virtual_id_number":773982,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-03 21:58:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-03 19:09:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-03 19:09:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 19:09:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-03 21:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c41c43b40ec241de06255bcc8b10894ec97be105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-03 18:40:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27b42619d464fa548031223987c0453d8a867a8f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-03 19:09:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLove it! Might even be worth having makeUrl return a proper URL all the time...","accounts_in_message":[],"_revision_number":1},{"id":"d727da973cdf02d9bad251992f4b30bf9e58497d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 21:58:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6fa8ee516b4d95b0697888d9eb15f8 : SUCCESS in 9m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef6373929a524d158c10a755ae51c1d9 : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/47fb4af2ce8e4c6f89eafcb1545de7fc : SUCCESS in 9m 36s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/936b4e608f7c461aa0e1aa63ed87628b : SUCCESS in 8m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f1af6b5055c6412cb8cfc86e10ff9817 : SUCCESS in 50m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa318e7497ba4a92891e7a161994de36 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"84916996fc22dfe6bacd1ac74d2750481218d128","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-05 19:04:03.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/533028","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f09c00fcebfc722a69259e7a036a8c3a39aac48","revisions":{"0f09c00fcebfc722a69259e7a036a8c3a39aac48":{"kind":"REWORK","_number":1,"created":"2021-02-03 18:40:39.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/82/773982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/773982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/773982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/773982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/773982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/773982/1"}}},"commit":{"parents":[{"commit":"04574c8fe4ac3ebc0415c482b90dcad4d2ba9279","subject":"Add random query string to CORS test requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04574c8fe4ac3ebc0415c482b90dcad4d2ba9279"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-02-03 18:40:31.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-02-03 18:40:31.000000000","tz":-360},"subject":"use some javascript builtins","message":"use some javascript builtins\n\nChange-Id: I17ef7c2d1a71479b7dca68a3a5c83b5a0e296e09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f09c00fcebfc722a69259e7a036a8c3a39aac48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f09c00fcebfc722a69259e7a036a8c3a39aac48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
