)]}'
{"id":"openstack%2Fswift~774195","triplet_id":"openstack%2Fswift~master~I3a5e05b91b29247eeb52ee88bb0a822f45f0523c","project":"openstack/swift","branch":"master","topic":"fill_all_gaps","hashtags":[],"change_id":"I3a5e05b91b29247eeb52ee88bb0a822f45f0523c","subject":"Make fill_gaps fill all gaps with own_SR","status":"NEW","created":"2021-02-05 06:20:08.000000000","updated":"2021-02-05 15:37:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"78f02afcb0faaae0b1c5c933854cf4d0d32ad23d","_number":774195,"virtual_id_number":774195,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-05 08:01:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 08:01:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 15:37:02.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"94c2bea9cb06a7ff84a7b5eaa153193d12df2850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-05 06:20:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6055ca13093ab972536910fd18ee01442cebcc1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 08:01:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd23146f1ef2489a8965f924967ddcc7 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/3465672acd2c46968cd2f4320bbe6802 : SUCCESS in 1h 00m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d58fc5fd6e724ea99c0f8eee624b4e71 : SUCCESS in 44m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78095555e9514f908b9e04f76d3b900c : SUCCESS in 11m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/44262cf515f14cca8ea93bdc6083851d : SUCCESS in 11m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ebd23db0b6dc44cf8df640effc9b64ef : FAILURE in 15m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e80fd0c08484200b1ef26cedf89903f : FAILURE in 13m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff292f162b5b4262b774c92584df2dce : FAILURE in 15m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/02ca167284e542e29528539eece412ff : FAILURE in 19m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/856bd611a7b2474883211bd5010f5b1b : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/159f2b28d2d84a8e853a532c1a934993 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a9c5c8ec919440ba5f6ca26c420a395 : SUCCESS in 30m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6da9bdec97bc46288d7ebb3b9ae5fa6d : SUCCESS in 41m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a01372cf6d334693b151ff56ac9ef9ca : SUCCESS in 19m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9d71e68db1334a3d97e83ed449d72940 : SUCCESS in 16m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e90713aa6324493fab701593b2ade33b : SUCCESS in 21m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/462704ad3caa48c185c21e30b7eba893 : FAILURE in 47m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3589c8025650413a9cfed534c48487a3 : FAILURE in 53m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a87d6e6f7b2740f5ba1b63505534f5fd : SUCCESS in 41m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43beac4be93840cbba91e1f34996d081 : SUCCESS in 43m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa66f8214fbd4fd6aa84c30e05db37d2 : FAILURE in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c486c1d0b95f407d8420e783a9207c05 : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e13506ecb2cb41d3bf34a5a53a3f4227 : FAILURE in 8m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/452fd587fb954145b62a6f4e86fde83e : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fafa87786c175980206a7021d0d1a79eb4122f00","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-05 12:18:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78f02afcb0faaae0b1c5c933854cf4d0d32ad23d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-05 15:37:02.000000000","message":"Patch Set 1:\n\nagree that fill_gaps is a bit lacking - it\u0027s more like \"fill_tail\"\n\nI suspect we may want to fill gaps with the next upper range that spans the gap, so that gaps can be filled by acceptor shards, falling back to root in the case of no other acceptors","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c5dfd49ff00c87a80346d982d0722e26d85bd90","revisions":{"8c5dfd49ff00c87a80346d982d0722e26d85bd90":{"kind":"REWORK","_number":1,"created":"2021-02-05 06:20:08.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/95/774195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/774195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/774195/1"}}},"commit":{"parents":[{"commit":"fb68315a3414bb3d75d5c8e689afffa372644a1f","subject":"swift-manage-shard-ranges: add \u0027compact\u0027 command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb68315a3414bb3d75d5c8e689afffa372644a1f"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-05 06:12:37.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-05 06:12:37.000000000","tz":660},"subject":"Make fill_gaps fill all gaps with own_SR","message":"Make fill_gaps fill all gaps with own_SR\n\nThere is a bug where we could mark everything to collapse back into the root\ncontainer and until all sharder have run any shrinking shard that runs\nbefore the rest will collapse and cause a momentary hole in the object\nlisting for the sharded container.\n\nWe currently have code in get_shard_ranges that when listing will\nfill_gaps with the root SR to plug holes. This is used when we are in\nthe sharding state and we need to still build full listings.\n\nThis patch makes this fill_gaps code work for all gaps in a returning\nlisting, if the own_sr is in a LISTING state, ie SHARDING, CLEAVED,\nACTIVE, SHRINKING. This way if a shard is collapsed back to the root\nbefore any others, the root will already respond to listing for it.\n\nChange-Id: I3a5e05b91b29247eeb52ee88bb0a822f45f0523c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c5dfd49ff00c87a80346d982d0722e26d85bd90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c5dfd49ff00c87a80346d982d0722e26d85bd90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
