)]}'
{"id":"openstack%2Fswift~774326","triplet_id":"openstack%2Fswift~stable%2Ftrain~I239884e0cf3e6438fc791d609fe3acd456e21e94","project":"openstack/swift","branch":"stable/train","topic":"train-sharding","hashtags":[],"change_id":"I239884e0cf3e6438fc791d609fe3acd456e21e94","subject":"sharding: Tolerate blank limits when listing","status":"MERGED","created":"2021-02-06 02:29:07.000000000","updated":"2021-02-12 03:49:59.000000000","submitted":"2021-02-12 03:45:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774326-train-sharding","meta_rev_id":"b2d8f0adb8a84e1f78022cf0101be2eb0e136976","_number":774326,"virtual_id_number":774326,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 03:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-11 19:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-11 19:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 02:29:07.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-06 03:33:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3adda112443e15cbf1ff0517f656a39be5985b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-06 02:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5943c9a585064eb3d345dfaed8a1692fc2304b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:33:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2efa90bdf9479c96d7363afe069918 : SUCCESS in 5m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4cd6645e2c2642daa1a5dce8e240b154 : SUCCESS in 1h 01m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2562fb76964e43858358dc03d15668d0 : SUCCESS in 41m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f6e69be80c8949798e495dbe5fcb2073 : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e056ae9dba64571b637833865db6a24 : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/38e7c81a27d34282a8c10745aa2d9441 : SUCCESS in 17m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/764139b6fcb440168c05cd70a5b0d67b : SUCCESS in 14m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bed487f76f64c22b87c6ce6ecf6f19c : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b32e98235b944f468e97ee1302eeb7a5 : SUCCESS in 12m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/30da956059084c7b916307877320a68a : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7aedabdd8fcb434c8e3a0899495e523c : SUCCESS in 11m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fa86987eed1a4ab996e4c1f6afcb00c8 : SUCCESS in 13m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3874efa2bb944c8985d95359ea7fea78 : SUCCESS in 9m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c147caa009ba4aff8799f0786ef3b9e5 : SUCCESS in 12m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9a79d0dc3b06474abc266bed043d4fba : SUCCESS in 10m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dd636bd360e84de684b033083640a238 : SUCCESS in 12m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/587d4902bc364f8f935df4f679ecf039 : SUCCESS in 14m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70afc6fadfb7446bbf6a948ed2d2d26e : RETRY_LIMIT in 3m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5a604209a1954d7a9e6d9d76584a3c48 : RETRY_LIMIT in 3m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf4a34f4994a4c71b2c335ddea4b2705 : SUCCESS in 28m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e2f6d0693b344c39b3d3e2f9e8d1a3b4 : SUCCESS in 27m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93a37cf286674b49ae1aab5c68fdb108 : SUCCESS in 29m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14cd95c24a3b447e8293b1b62a860497 : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3747f3d679a4cd6bcd30f82c863981f : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3704d2805a964f5ba031391d3c00bf87 : RETRY_LIMIT in 6m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8aa9f3e8595a4862b5d5ac7b0840a8b2 : SUCCESS in 49m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/09e5681c941544a680e0573bfeff97ce : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cabead553a72dd5671dbe9b769e5a4a9ac1c4361","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 17:21:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fdb2bac23019c016b56223c2777b0352bf0cf07f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 19:44:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, fixes a 500, LGTM!","accounts_in_message":[],"_revision_number":1},{"id":"697dac403fbcd6378b1ce0f54737f224fb3e2894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 22:10:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0dd013cf7f4c6789891bc31105b83b : SUCCESS in 5m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a8a6f5fdcfca4ec08c08df2b0473fa93 : SUCCESS in 1h 01m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdffbe3eac0349c092055abf7879adc1 : SUCCESS in 37m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0992c5c54b7f4390bdd4ba3e78ea1be5 : SUCCESS in 8m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3d2a35670e9c45bdbb5ecdfb2c01a491 : SUCCESS in 13m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/85b1f3e297934f5e99760a96d14627ba : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/006ad90918b74f759af57f714dc26751 : SUCCESS in 11m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3554946fcc3043569daff5654ef45724 : SUCCESS in 11m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/319e55af9af54d76a0155f4bef28c198 : SUCCESS in 13m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/aaba5609d5f246938a9b525d9687cffc : SUCCESS in 10m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8e4f430ce7e6414a980cca836f363427 : SUCCESS in 10m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c44580c8954d4ef4a09bb268ff2317c7 : SUCCESS in 12m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8f7aeb964d074111945197a51046147a : SUCCESS in 10m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/68b9eb40159f480db4348d560772b743 : SUCCESS in 11m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/66601768d0f94923b10852b11939fd10 : SUCCESS in 10m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2b4664f538184e92a41e93577e37c9f9 : SUCCESS in 10m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cd67fa1a150b4cdca37c30bb0180e93e : SUCCESS in 13m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2ace8d6081b4507af5e658e0df70921 : POST_FAILURE in 5m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b4d72e55c71145ba9532f8631f952936 : TIMED_OUT in 1h 01m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff30a5667e2e461ab0056f1e65cd9950 : SUCCESS in 31m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/332e0b97f688473592fccd2274de0ed8 : SUCCESS in 32m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9e8fee51df74b1fabb9ceb6efed8b91 : SUCCESS in 29m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c5bbe52a17a47c2b8f77b283841219c : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af187779da5c41c4a966cbb100e8d604 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ce8a37d41fb4937a60310240392af3c : FAILURE in 7m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3368ded54543488f90b4aa939680438e : SUCCESS in 1h 01m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7ab67efab6ff49658a40ca35055b8bc2 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":1},{"id":"061293915bf1f65328ec40c2efa9b913dd0ced83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 23:21:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b12a000873699c408f43fb1c0c21ec7f5fde64a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:40:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0f0b71189c498d9a121ded5e27cfac : SUCCESS in 6m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/e3e8e0fbe844444ca9fc3cae1a9eabc8 : SUCCESS in 1h 00m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9618ff31bdcd4b1db420a3c4dad27c60 : SUCCESS in 40m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c26437b5428048f6a7ddbbb3d9248b1b : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/742d0e6e45604450ac5319bd0f36a9d8 : SUCCESS in 21m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ee635397c844eac9162ba0b1cb10e39 : SUCCESS in 20m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/10b995693b0b4b67acf5cf601e9e0768 : SUCCESS in 14m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75b131b85b09441cb1506cfa09506d69 : SUCCESS in 10m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/565c9178b3a646b69acd275430390de5 : SUCCESS in 28m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1cae3f771048420a82f067cf129551ea : SUCCESS in 27m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/79abb14865b1419c8db492f151412a18 : SUCCESS in 10m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a12cb75c8d394cd8b4f72ed08458b164 : SUCCESS in 12m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5cdb06ef6ed441249068adf5cd6b758d : SUCCESS in 10m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b603833bca104fff9cd331ca7d5b14f7 : SUCCESS in 11m 20s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/94c054e6010f4f96bb7ad417c6c73dd3 : SUCCESS in 9m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c82fbb9c550b4a52b92ff0ddfefc7168 : SUCCESS in 12m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/44e5f89b4d894cac88cc180c3b7bec6c : SUCCESS in 19m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da09c30609b14e68b2a3db46bec5ea2e : POST_FAILURE in 5m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39867842581d4c9f97e6050fef376c78 : SUCCESS in 29m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a99e205f89d84e89b3897130a4d6017f : SUCCESS in 28m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e6be2bad2bbf4718aa4f545ceb2709e6 : SUCCESS in 28m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3df267de3314ebd93c25a5319099bbd : SUCCESS in 36m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a554ecaf11334cae9f8611d757e17a2d : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/253709854677454c8ca68981b738b8e8 : SUCCESS in 7m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/477a3d95869a42a987a9a932a303d86e : FAILURE in 9m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/901919b0589c449496fe79c4fcd1f86b : SUCCESS in 52m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/045dfe89f5bd482f866c48035ede22a9 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b33ae66532d3a29cc367b6c78271e01983f81257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:41:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2969c60ee29be9daa9b1ade7e6ce864bc76cef53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 03:45:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9394fa5166394781bbd98ac804763e8c : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/df9d8d8d2aca4fdf8f8e14f2e8842a1e : SUCCESS in 46m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ccbbe4046b747dca205862a776e7914 : SUCCESS in 41m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0d09ec837e4945989d53afaf2c15f6d1 : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d70f6bddba247d699f5077c634e9295 : SUCCESS in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/399ecf25eed64992894c568e24e08fcb : SUCCESS in 14m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50b0eff5a5564f7380abeaf097beb769 : SUCCESS in 16m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/031f4b327b774375a5a6f7c94cfedf3b : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13a0da10ca834de4912dc42a56d90baa : SUCCESS in 10m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/168d75dc2ac84b17b49f38c5cb84204e : SUCCESS in 9m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d059a8f0662b452399c8dd525e2150ab : SUCCESS in 17m 34s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d3d68690c67c48e1a61ce336b87566b8 : SUCCESS in 17m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/44e961de5e9a4ef283c8ea6434956a41 : SUCCESS in 19m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6f990f01f8144d03a6087c286fab0c38 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/76d721614068458698f7e9c1a22d49ee : SUCCESS in 15m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4194c38a148747b0af19ae458b2ac18a : SUCCESS in 22m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/76f409aaf5f84649821f151ce0eda202 : SUCCESS in 35m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a011af561b94236b66ae4b1365c02d7 : SUCCESS in 32m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/519a92d18b3c404aa8c02a356f0725ca : SUCCESS in 32m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/abb422da34fd4073b0e1366ece8be9db : SUCCESS in 24m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/114f192e687b4d7ea1f0b91e681caf80 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b286929ff2d446b99f92b932b80ca4b : SUCCESS in 6m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2270129f4d74eeabd5ad89420e8a183 : SUCCESS in 50m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e4a3f99ce48b4b70a98d9ca33cd71fa0 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2ce81c549976ae5ba2f21d4c73f9012f69614e4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 03:45:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b2d8f0adb8a84e1f78022cf0101be2eb0e136976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 03:49:59.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f0d39717314243afa1f55cdb312249 : FAILURE in 1m 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bd56de5fd26341aba43bad28f1e59bfd : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9ca078e70d48f3292a89dbfe7579e7e7d070d97","revisions":{"c9ca078e70d48f3292a89dbfe7579e7e7d070d97":{"kind":"REWORK","_number":1,"created":"2021-02-06 02:29:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/26/774326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/774326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/774326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/774326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/774326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/774326/1"}}},"commit":{"parents":[{"commit":"d63b7013a8e363200ad5c2c5c039b37aa7881878","subject":"[stable-only] Cap bandit to 1.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d63b7013a8e363200ad5c2c5c039b37aa7881878"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-20 06:27:27.000000000","tz":-480},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-06 02:23:57.000000000","tz":-360},"subject":"sharding: Tolerate blank limits when listing","message":"sharding: Tolerate blank limits when listing\n\nOtherwise, we can 500 with\n\n   ValueError: invalid literal for int() with base 10: \u0027\u0027\n\nCherry-Picked-From: I35614aa4b42e61d97929579dcb16f7dfc9fef96f\nChange-Id: I239884e0cf3e6438fc791d609fe3acd456e21e94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9ca078e70d48f3292a89dbfe7579e7e7d070d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9ca078e70d48f3292a89dbfe7579e7e7d070d97"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
