)]}'
{"id":"openstack%2Fswift~774327","triplet_id":"openstack%2Fswift~stable%2Ftrain~Idd634d1152395a50caa59f35452f216007e08406","project":"openstack/swift","branch":"stable/train","topic":"train-sharding","hashtags":[],"change_id":"Idd634d1152395a50caa59f35452f216007e08406","subject":"sharder: quote() more Swift paths when logging","status":"MERGED","created":"2021-02-06 02:29:07.000000000","updated":"2021-02-12 03:46:48.000000000","submitted":"2021-02-12 03:46:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774327-train-sharding","meta_rev_id":"710e3fcabfebdb92153d3673052cb0fedbc58e7d","_number":774327,"virtual_id_number":774327,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 03:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-11 19:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-11 19:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 02:29:07.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-06 03:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e7a1bcb061014c9639e1fe5fb9685e1e4355990","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-06 02:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a76251067b1aa6b3ae38e2af8f5270e4eacbe7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:38:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4198cc260f514d3fabd66bed86d47e6d : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/99e1ccb2cc0241a0a65331324c705963 : SUCCESS in 1h 03m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/653343a168414019bb51e344379f4bb8 : SUCCESS in 48m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8304bbdcdec4431d9227fec97a8b2c10 : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f9934f0cfbd45e792358313aee41402 : SUCCESS in 13m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e60f239178a548bba0620c468107989f : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6f49328559ce4573bca3d5c7d11060ba : SUCCESS in 12m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/833a9bb810ad419c8b2e119a8c68eaf3 : SUCCESS in 9m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a64ffcb2e06c4a99b31f974e7bd51d03 : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0cb12ddf9d9c4f469e4a24386541293b : SUCCESS in 9m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c63a1d4441684a3dab3550af2942f165 : SUCCESS in 11m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c57929b2f1904e9ab8c3bd19c29c9e25 : SUCCESS in 13m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/64231f2ddc5d419b8b1ddb1f28a3e16e : SUCCESS in 10m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0dc30f1805e94378895f85196075998c : SUCCESS in 10m 28s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4b2ca13ef445426982cffd57e22727da : SUCCESS in 10m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d5253394f81e49e1b48edc292b97ed2b : SUCCESS in 11m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4fdbf43ed3c144f092d17e2d810d0794 : SUCCESS in 13m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1fda659f66be41beb0edcc3fb6c4eaed : RETRY_LIMIT in 4m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3f5b135b3faf44f5bdb8c6b330fa2903 : RETRY_LIMIT in 3m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/033dd67f25614b1e8eee8bb6ef7d22ad : SUCCESS in 29m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/faa83c7a6d6c49f281b9da19524ca617 : TIMED_OUT in 47m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89057321aa864c3eb735bd4666d58dd6 : SUCCESS in 27m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afc16a7b32fc4148b1bfd435767b1c56 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3012ffef794fd8824021ac55966f2c : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a42161a8a4e4004a2b4968b7eb58cfc : RETRY_LIMIT in 5m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3838cbad2fb04176905711f634876836 : SUCCESS in 57m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dad05826b93e400cb1b914f6b7edd9c5 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6e895c1224f10544a599f7409ea4950b35a25b25","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 19:48:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, patch was present when ussuri was cut. I probably should have included tests, but oh well.","accounts_in_message":[],"_revision_number":1},{"id":"cd79a9d0e68c2bbf8a6a487958dc9718a65b93f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:14:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bb5c841e484df587a278cd547a9157 : SUCCESS in 6m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e518fbcef6894babb89d4ba54f08f565 : SUCCESS in 56m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/341ce07d836a40389615570f7a93f633 : SUCCESS in 49m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbeddd83052649258b69105941c98281 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0a8e7006b284bab8bb76f911214db63 : SUCCESS in 10m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ca5b64492764267b829e071d73c16cc : SUCCESS in 10m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/17071ffb037945acbf4d2291d57aac25 : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec33d2ea88c94a3e9a83e9d63ece1d05 : SUCCESS in 10m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d9b5b5804833445abec820b7d8858c5c : SUCCESS in 11m 26s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/481a37752ddd494aa00e142a3a48731c : SUCCESS in 12m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7f6678d45503485e96d41a0ba8d0e0dd : SUCCESS in 11m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1cbcaf5f228d47c084c939e85058f846 : SUCCESS in 12m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8995885cb3a34d6ea8dd9edc919b4a8e : FAILURE in 15m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5159bec6fe344aca9d07870400acfba3 : SUCCESS in 11m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/99838fec3aed47aa95c6178b11e0808e : SUCCESS in 11m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6835076b3f2c4a2a81fdf5aba9fcc1f8 : SUCCESS in 10m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b07c351c32d14f99b32afccdc5d2f704 : SUCCESS in 13m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f5a65b1368b4ab6987b9b01d8cbc358 : POST_FAILURE in 7m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/05ad31bb2c554d09ad4c638ea461fa9e : SUCCESS in 34m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca757912546a44e391020ffc4b9d1021 : SUCCESS in 32m 19s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/24d4b89cdb924c1898b3fd9d36035682 : SUCCESS in 32m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b34cafb56504b599d0ee4978ed88e38 : SUCCESS in 27m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/700863a8d4f04dad98d49cb1d6da8d9c : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5fd7fa342f4c06baeeb9084c21894a : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ee5838b578549ceb4480261b17d018a : FAILURE in 12m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66023dddaf5c4dab9050505c21181061 : SUCCESS in 54m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/092129d4ea9846bdbb628f279fe17dd9 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2934a7c419d735c4e8447ec591cf45e4d72b2288","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 23:22:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4f1626fe5542f3dca1ee47a4d2ab02ac6cfe8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:50:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f632c9da73a840f69f0407bda9851bf7 : SUCCESS in 5m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/1270a3b4d3b24519ac9cc24ce965dabc : SUCCESS in 1h 04m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f960776e0ab6490cab2637e59ffea9b4 : SUCCESS in 46m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcdc7df0c8ca48aa954e139e354efd1a : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58f7b6b382954aa8b08931934e14136c : SUCCESS in 11m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73a7c94d25bc4b7ca5cade66512de22e : SUCCESS in 12m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/46f698d8b32346cf916e413e92e3e1ae : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/162190c798fe4891b842da4a2e9b238c : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f72b6aaf12764ad3b58471785330ed73 : SUCCESS in 12m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/99678f15f3934279b0e3d3ed9648c9d3 : SUCCESS in 9m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89da9ea799f44bff855f9448274276d6 : SUCCESS in 11m 53s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/aeaf6f6e7f684f11af95ae4081dda982 : SUCCESS in 12m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/15fb1107937846e196380bc3508d0a40 : SUCCESS in 9m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1749a09cf0e7469c97e1e49e5d098dd1 : SUCCESS in 10m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c49f0b9085c748a9bc2d13cb5155e37d : SUCCESS in 12m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a77c30c6dbf64851bf8172f486bfc8a0 : SUCCESS in 12m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/33390d00e3934ea3a7d17f6ec7ac2c75 : SUCCESS in 15m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d131187c63ad46a98ceeec88860c3112 : POST_FAILURE in 5m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/26e999471c9742b2b91ed90a7998d84a : SUCCESS in 34m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/559153eb1caf413c8b61d146c32f6235 : SUCCESS in 31m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ef7b2e419d5647829239d19561ddb12f : SUCCESS in 33m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9b79896f6174d7a964d17e984586d41 : SUCCESS in 30m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e77a1272155140fd8fd62f03218a890d : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c988616cf24972b51384891ec638c9 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/148c9ce09a044c9ab51dcfc10840aa39 : FAILURE in 11m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ba0944c4bbc452cadc896c87c578cdf : SUCCESS in 1h 01m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/72bc4f3c4a514aafb825019a85036734 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9c874eefeee1ff4a4d5f0efcbf9c62f749b88e78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:51:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e79ab7892eb65d65ffa55d94687b798c06ab4a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 03:46:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeec7840dce545b18f43ce0219f298b9 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/8e63bda39d544a7d9ed42157100f208e : SUCCESS in 47m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c6d90ea22d64bbe82dd991da7dff1a6 : SUCCESS in 46m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cbc652334bb64a4fb50f07aea8dd8857 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d53a31c1332b48bcb8a3dd71dc633c06 : SUCCESS in 12m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01336c06e7734d82a0cb96071ad10ae9 : SUCCESS in 12m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/725f47b24b354fcc96acc0b840f0886b : SUCCESS in 12m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d5d4dbc2f97640f5bdc9a541b9b9e009 : SUCCESS in 12m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4799eccb8aac4db5aecc7af0e05b0f95 : SUCCESS in 11m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/872577a243df41faa6ecb5de008b1e9d : SUCCESS in 9m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10c9fc8572ce45d6b77d75b98bde8963 : SUCCESS in 10m 40s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f53bd931735443e596ca29eda52d5dcf : SUCCESS in 12m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dec6fa67994e4751b5f1923d56b637b5 : SUCCESS in 14m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/97939cfbba1c420b98b33f6aa8870c46 : SUCCESS in 11m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e93c866f92ed4120bb97828cda3dd903 : SUCCESS in 14m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/27d7ffbce3b1480c914bb71e963b4264 : SUCCESS in 13m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/478d51efce924b84b4207fc043042154 : SUCCESS in 29m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/091af843dbd1425a82a14b852367df6c : SUCCESS in 29m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ab97fd177c3448b880203c5820f151b0 : SUCCESS in 30m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f4e67cabc8b45188c57529e4b091367 : SUCCESS in 35m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/883bd2a7fb8e48c5849418082d4f2619 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd166d20f424751bfc2b2b7bc7068f7 : SUCCESS in 5m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d23c1d5e391454cbdfcc2492adf2b50 : SUCCESS in 50m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b738a6d14268483981b29799ed9502ff : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":1},{"id":"710e3fcabfebdb92153d3673052cb0fedbc58e7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 03:46:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a6ffd33fbc9aa8fb0d0d45063c4701810c70bd4","revisions":{"2a6ffd33fbc9aa8fb0d0d45063c4701810c70bd4":{"kind":"REWORK","_number":1,"created":"2021-02-06 02:29:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/27/774327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/774327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/774327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/774327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/774327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/774327/1"}}},"commit":{"parents":[{"commit":"c9ca078e70d48f3292a89dbfe7579e7e7d070d97","subject":"sharding: Tolerate blank limits when listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9ca078e70d48f3292a89dbfe7579e7e7d070d97"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-12-20 06:54:13.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-06 02:23:57.000000000","tz":-360},"subject":"sharder: quote() more Swift paths when logging","message":"sharder: quote() more Swift paths when logging\n\nThe lack of quoting gets extra troublesome with reserved names,\nwhere messages get truncated.\n\nCherry-Picked-From: I415901d3a8cd24cb3cedc72235292bb9d1705bbc\nChange-Id: Idd634d1152395a50caa59f35452f216007e08406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a6ffd33fbc9aa8fb0d0d45063c4701810c70bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a6ffd33fbc9aa8fb0d0d45063c4701810c70bd4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
