)]}'
{"id":"openstack%2Fswift~774339","triplet_id":"openstack%2Fswift~stable%2Ftrain~I9e3b6391b3ea85ecd147916a2bf0f002c35fe5e3","project":"openstack/swift","branch":"stable/train","topic":"train-sharding","hashtags":[],"change_id":"I9e3b6391b3ea85ecd147916a2bf0f002c35fe5e3","subject":"container-sharding: Stable overlap order","status":"MERGED","created":"2021-02-06 02:29:07.000000000","updated":"2021-03-01 10:25:22.000000000","submitted":"2021-03-01 10:25:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774339-train-sharding","meta_rev_id":"1b8e6d65061ac86cc53f4a05c31069bffbb315e0","_number":774339,"virtual_id_number":774339,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 10:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 23:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-26 23:12:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 02:29:07.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-02-06 03:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 22:41:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a5191f0dcc621099191871e92abda549cad99bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-06 02:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"486eda190ea0e49cb1831b1de0e0d95229ffb196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:46:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf53a6816a9486887ebd249e89f6d26 : SUCCESS in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/aca24995dab2494b8250cb8f0b72ee21 : SUCCESS in 55m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c72f19ec446345568a54daa0aaa26c8c : SUCCESS in 46m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/062b9ca5ab4a4c668e73c3c96d1963a6 : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/137a9af578f24aedbabff14a5ed19cc1 : SUCCESS in 12m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3eec8658b254e70909585b93ac89c96 : SUCCESS in 13m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7660b826df034b16bf2ba0de7ec9f33e : SUCCESS in 14m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71e7e3e94c634697b848d053e9f98ba3 : FAILURE in 10m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b64dfb47b97040ac8d75d3de9862503b : SUCCESS in 11m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/412d266148dd4cf88c8a7928ecd940f5 : SUCCESS in 11m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b61551d81b7342a3941a5dc70e3d4fde : SUCCESS in 9m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/12bdecb7e9a74bf09e1fbf769eff0e45 : SUCCESS in 12m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e6260d187d624a80a219731e8d7a4626 : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/857841fa2a4f425e86432875668ee1d1 : SUCCESS in 10m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3177e67180d04fd698ed43ef904c48c2 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e466dc7cadd2447cbc89ee1f4ec65f23 : SUCCESS in 11m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4aa89fc613ba436da0177b8e536532f4 : SUCCESS in 14m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9458bab099b49f885daa4c58222c5a4 : RETRY_LIMIT in 3m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3996fb9f101544eea9a8c909ce6f7e59 : RETRY_LIMIT in 4m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/610469ada4f84a769deceb4f423968b9 : SUCCESS in 27m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ef4ad594f9ce4b90900a56dae7b51fcf : SUCCESS in 27m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3410896a5354c83a63e70c313d2d9c5 : SUCCESS in 30m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcc9a614616f4d2b802b708b9af75c4a : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53c39386475c4bd48bd7b0c7035d0b6a : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45286152582d4d0e90e056cba054da8a : RETRY_LIMIT in 5m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40e6584020ba4c418f197f7c673c4424 : SUCCESS in 1h 00m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b5738fb61377480da70664c48f40183f : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4d48ff4a2fecc8423dedb556dbd28c728e55f0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-25 02:02:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a0f0720c51bcf4cc7e22716f9e0e143415de5612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:44:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4198203e3b244558945fe93f3e864fd2 : SUCCESS in 6m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/b4c91e9e769f487fa31270b4885aea28 : SUCCESS in 59m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5b65cc1a64a4272968cd779cf2f9181 : SUCCESS in 43m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a696b57c96a44441b038031b0a4ec8d5 : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a143536757304b6fa158acd81df56feb : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df04852f7f314d1f9d57e830f4de0f75 : SUCCESS in 12m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e29b09adab6b4a60968be2b8aa919503 : SUCCESS in 19m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0028245a88a14ec4902b7da5f1938bc1 : SUCCESS in 11m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/53cd7cacf073402a885c13a7d1066fb8 : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a486bd66ffc641d7b55a2eab05f32c93 : SUCCESS in 10m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e478a3896faf45daaead17401b24e64e : SUCCESS in 10m 57s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6f99cf6c96b943598c9574cfe663a423 : SUCCESS in 13m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1790c11bff66449f986bbde11aae3bc8 : SUCCESS in 10m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f02adaeb0ce54f758567fb886f781d62 : SUCCESS in 15m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e7b0940d011c4d86841d51a1f729f6de : SUCCESS in 11m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c29040f0495149409299614422bc5214 : SUCCESS in 11m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8a6a96e88f92499cbabe0f1e62ef7175 : SUCCESS in 15m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7b973883b5e94b918017aaaefb8155cc : SUCCESS in 22m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1904e8c788a64766a5545fc38a986c98 : SUCCESS in 1h 05m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9522febdfe1a48a4bfbe9957e3db72c6 : TIMED_OUT in 1h 02m 38s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/28219ebfd103403aa384b9fbe36f7cd6 : SUCCESS in 30m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/096b8cf51a2f40698c8f781114c292e9 : SUCCESS in 31m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4889e3652074880b760b2bcfb6e4261 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c3e57965fe4e00872a97d57832a077 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/045fdd15f6bb4d32ad186a31b7e84598 : FAILURE in 9m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f7c589e2344465d8d78ab2c59d1f716 : SUCCESS in 42m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/86e5106a61534a80b37929a35a8362f0 : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c6c58a57da38624dc91c6922dcb76886c42727b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 22:41:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"db8c43a6aafbaa66dd498f7ba991cbd398821622","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 23:12:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a7ccb41534f8d67c5c9d069fc96ea14337fd393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:45:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/316eff86832c42648b94f93bb08c0b0a : SUCCESS in 5m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/529259535ac84073897d894786c5c790 : SUCCESS in 58m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b62ae6d6aea4b658c331803ac3cb387 : SUCCESS in 44m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad298eafc9894486b447968d9e4ce119 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a9449603fbe459dac344137bb29ff95 : SUCCESS in 17m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/340f250d59f243a1b528eea9bedbbf24 : SUCCESS in 12m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/82c37c3c7e3b45a7bc4a6ca76ba98332 : SUCCESS in 12m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/918a14df87c849fc9609ae21ef1470ae : SUCCESS in 8m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bf44b8f538c24e9bb0612311ba84c8ef : SUCCESS in 9m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b1f820ef815843159924442c106f5976 : SUCCESS in 9m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d3696bed15534d108dfbf0a1454aa4dd : SUCCESS in 10m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/302f7a2041c44a758ebc43af6242405d : SUCCESS in 12m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3ee63470691747e285d02369733f484e : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/76390a85bbfe4285ba7898c2bd499fdb : SUCCESS in 10m 34s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ffcd731de77c4b08b7cd26a7787cf335 : SUCCESS in 9m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/58456716e84e48e698c356ce31ca51b4 : SUCCESS in 10m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/62842a4239754a61926e19530693b986 : SUCCESS in 12m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb42ea8915fc4d95a8e28ffa71b64ce5 : SUCCESS in 20m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/069f5f64768647359b79cd7e9663fffd : SUCCESS in 36m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/107da73a529d448e8b1ae1014aa3a2cb : SUCCESS in 25m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/afb9bd1f95c042aeb612502fddeb8fd4 : SUCCESS in 27m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a32d0875e6e4b2bb620456d7b1a4e38 : SUCCESS in 27m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8531d3c7e9844211b6ae3dc7db603eba : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/686cc3c0da7c44788be2c9d4a5145af3 : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d0d85283adc404e890415979ad05e2b : FAILURE in 13m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1a0e13f104d46548d5196ad3eb43f4b : SUCCESS in 52m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b36f2fe2a1b24275bab22ebeb49f7110 : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":3},{"id":"5f4edd93b94e78e9d91e22f29adca44e4084279a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:31:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8278c78a7b12f98b04ef4539171a4d5e1e8a5d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 03:34:27.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"bd1a70cb08ff694ad9f386d2869d537445e59127","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-01 06:24:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"16ffdabd52b1a04f517390e2d9a41fb1dd09ade0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 08:04:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9114833b9c8d47e7bedb0b88211845ff : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/fda1281c9e1e4843a8966cb44f8c2d64 : SUCCESS in 1h 04m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8889fe52ecea4e76ae393bc2e903c6cd : SUCCESS in 44m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9a5d049a4334edea9bc11e5696a94e6 : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0715efee583453296b1670f7b3f8621 : SUCCESS in 11m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81f03ed650d6439c9dcbea51e83152e9 : SUCCESS in 12m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a42a02430e04416826efda381d1ebfc : SUCCESS in 13m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ecddccafbe32401789f94678dd985362 : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4241dbf52c1546f6b12cfb168f7a980d : SUCCESS in 9m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8b06dfd31fc5453d889e9fde716c89f8 : SUCCESS in 9m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da979813860f46c08ee4e24c52547379 : SUCCESS in 16m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/016f167544d14a02815f4c9bb2ad99b4 : SUCCESS in 14m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b48b6949a84f4af4916c5694872f7fc2 : SUCCESS in 10m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fcfbc4c7dd3840998556315dfed4e36c : SUCCESS in 10m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e6efc0136574444e8a2c34312254f7a2 : SUCCESS in 10m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/37aebd2d63ac478fba97fdb17cacd7b4 : SUCCESS in 12m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4e965b4490c349078768f55ba818f0fa : SUCCESS in 14m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b6c1a8081054d42b31e9d805fd3b1a5 : SUCCESS in 21m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9393c5fd78cd49d9bef8a1120de8071f : SUCCESS in 38m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d80627607b344d529a40ce720794cc19 : SUCCESS in 28m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/aa4269cda61644e7bc94ee007d542955 : SUCCESS in 31m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27a8f7b8d6a443119d3d3672860c9b16 : SUCCESS in 28m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/262e204822034ff5bef97cb6ac97b064 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c29b165dc1442389766ca828d528c28 : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0e31505a3fb4dba841dc0c10ec87a42 : FAILURE in 10m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed167f0b6b7a427e969dc6357397feb0 : SUCCESS in 56m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f882085e0043475d9794f34a90702624 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ce5665974376f1c3ea100a5eadec0574052dec03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 08:10:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8d29c43076eb9ed30802453567bddb37df05bf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:25:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1cf9b4ecb9c4bcfa266dc3601489db6 : SUCCESS in 7m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/cbc066cf4fd9450c93ae2dc94cb7d351 : SUCCESS in 1h 05m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c9e16f1f1c24d58a0366fece2cfa06e : SUCCESS in 48m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dc5c14fac72b4068bd26327b4b9b0492 : SUCCESS in 12m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17505879ca9049f09503895e80348716 : SUCCESS in 15m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c375d4704aa44b8be4aedc4e725ffa9 : SUCCESS in 14m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c95b139bb99e40f491f86b607464c518 : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7978ace6463149109818f5c6bcf0e16d : SUCCESS in 12m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21d210e36bf44937b667a9c396b57e27 : SUCCESS in 12m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/52c401f9f23548bb80c7dfa55dba6aa1 : SUCCESS in 12m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a18944bb98f14bb88ef1cff272437925 : SUCCESS in 13m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f207ea5400a04b4aac10e1be0522dc74 : SUCCESS in 16m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b2c0c46835d8472fb9e5313d68a5291e : SUCCESS in 12m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7b60f44e39a343e5a6bde79dfb55c490 : SUCCESS in 13m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3db8a920e0e8429f88ae34d6ff15369d : SUCCESS in 15m 15s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ab4c110d914048ec8a5e125bcf2beece : SUCCESS in 16m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ceb10b6da9ad4b2a8e4e1e93e423ea29 : SUCCESS in 44m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2f4f99580f04201a0e8ca329d13e3f5 : SUCCESS in 30m 15s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cc505ebce8d44be0b7345fde7e77c9b8 : SUCCESS in 33m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4ae53238080493a8396d6058eb76cdd : SUCCESS in 29m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1ff5badf70c4204884ae4d78a1a7383 : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43df8ce9bbb431cb6e96e0ccf259874 : SUCCESS in 6m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4ad91bfb7eb459182ee6df54b81ff1e : SUCCESS in 56m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ef0f36060b704e04ac1385ce133c2714 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":3},{"id":"1b8e6d65061ac86cc53f4a05c31069bffbb315e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:25:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"71921f8612e1de682f96962ac1a7094f789dcff1","revisions":{"743f4f0cd31740568f3c5dc65e87b007da93e074":{"kind":"REWORK","_number":1,"created":"2021-02-06 02:29:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/39/774339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/774339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/774339/1"}}},"commit":{"parents":[{"commit":"e849442be2e00295e2ca4fd61023c68c9394f791","subject":"Address a sharder/replicator race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e849442be2e00295e2ca4fd61023c68c9394f791"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-22 20:17:08.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-06 02:23:58.000000000","tz":-360},"subject":"container-sharding: Stable overlap order","message":"container-sharding: Stable overlap order\n\nWhen you have overlapping active shard ranges updates will get sent to\n\"the first\" database; but when the proxy queries shards for listings\nthey get stitched together end-to-end with markers.\n\nThis means mostly the second shard range is ignored.  But since the\norder of shard ranges is not stable (it seems to depend on the database\nindexes; which can change when rows are added or removed) you could send\nupdates to \"the wrong\" shard.\n\nUsing a stable order leads to more correct and robust behavior under\nfailure; and is also better for cognitive overhead.\n\nCherry-Picked-From: Ia9d29822bf07757fc1cf58ded90b49f12b7b2c24\nChange-Id: I9e3b6391b3ea85ecd147916a2bf0f002c35fe5e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/743f4f0cd31740568f3c5dc65e87b007da93e074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/743f4f0cd31740568f3c5dc65e87b007da93e074"}]},"branch":"refs/heads/stable/train"},"469b17902835ac5658546931802e3666e7f0f21b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-25 02:02:32.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/39/774339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/774339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/774339/2"}}},"commit":{"parents":[{"commit":"3cce55dcee55070c5b5bbefdd912d47e84f4f4e9","subject":"Address a sharder/replicator race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cce55dcee55070c5b5bbefdd912d47e84f4f4e9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-22 20:17:08.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-25 01:51:53.000000000","tz":-360},"subject":"container-sharding: Stable overlap order","message":"container-sharding: Stable overlap order\n\nWhen you have overlapping active shard ranges updates will get sent to\n\"the first\" database; but when the proxy queries shards for listings\nthey get stitched together end-to-end with markers.\n\nThis means mostly the second shard range is ignored.  But since the\norder of shard ranges is not stable (it seems to depend on the database\nindexes; which can change when rows are added or removed) you could send\nupdates to \"the wrong\" shard.\n\nUsing a stable order leads to more correct and robust behavior under\nfailure; and is also better for cognitive overhead.\n\nCherry-Picked-From: Ia9d29822bf07757fc1cf58ded90b49f12b7b2c24\nChange-Id: I9e3b6391b3ea85ecd147916a2bf0f002c35fe5e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/469b17902835ac5658546931802e3666e7f0f21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/469b17902835ac5658546931802e3666e7f0f21b"}]},"branch":"refs/heads/stable/train"},"71921f8612e1de682f96962ac1a7094f789dcff1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-26 22:41:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/39/774339/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/774339/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/774339/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/774339/3"}}},"commit":{"parents":[{"commit":"87c1aa0e0701a35f7bd16913c7622f1445081423","subject":"Address a sharder/replicator race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87c1aa0e0701a35f7bd16913c7622f1445081423"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-22 20:17:08.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-26 22:39:05.000000000","tz":-480},"subject":"container-sharding: Stable overlap order","message":"container-sharding: Stable overlap order\n\nWhen you have overlapping active shard ranges updates will get sent to\n\"the first\" database; but when the proxy queries shards for listings\nthey get stitched together end-to-end with markers.\n\nThis means mostly the second shard range is ignored.  But since the\norder of shard ranges is not stable (it seems to depend on the database\nindexes; which can change when rows are added or removed) you could send\nupdates to \"the wrong\" shard.\n\nUsing a stable order leads to more correct and robust behavior under\nfailure; and is also better for cognitive overhead.\n\nCherry-Picked-From: Ia9d29822bf07757fc1cf58ded90b49f12b7b2c24\nChange-Id: I9e3b6391b3ea85ecd147916a2bf0f002c35fe5e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71921f8612e1de682f96962ac1a7094f789dcff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71921f8612e1de682f96962ac1a7094f789dcff1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
