)]}'
{"id":"openstack%2Fswift~774343","triplet_id":"openstack%2Fswift~stable%2Ftrain~I3a197fecfd68359a325efdfd4590b7bd48598da8","project":"openstack/swift","branch":"stable/train","topic":"train-sharding","hashtags":[],"change_id":"I3a197fecfd68359a325efdfd4590b7bd48598da8","subject":"Avoid loops when gathering container listings from shards","status":"MERGED","created":"2021-02-06 02:29:07.000000000","updated":"2021-03-01 21:32:53.000000000","submitted":"2021-03-01 21:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774343-train-sharding","meta_rev_id":"991372c52124f8cb92918c3418e9d8cff6e55863","_number":774343,"virtual_id_number":774343,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 21:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-03-01 21:32:53.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 23:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-26 23:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 02:29:07.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-02-06 03:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 22:41:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a0c9e5bb7863b94dcedb2f5d49b4d427b6b5edfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-06 02:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d79f2400bf677221801dba02d0b6989f675723c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:53:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e8729fe3aa48b79b5b33b600aea8f5 : SUCCESS in 5m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/d03478d080544e6088d915625e3575da : SUCCESS in 58m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db1b46c647e343d18c34f7ce728d3915 : SUCCESS in 40m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70d1e01391c9455db05f51d81ce014c3 : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/435f1262de4b43f2858fcc0ac65be29a : SUCCESS in 12m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e4d70d7f7d5347319401b23734e47eda : SUCCESS in 12m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0589ac87c3d64ca885703c8f641b733d : SUCCESS in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5589a73b142e461698f83b51113cfa8b : SUCCESS in 9m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d71a9e67e5de487092f8dde3a33bdf94 : SUCCESS in 10m 10s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/719cdd7bb4e5453094ad0e0039177bc6 : SUCCESS in 8m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05a2415878e54001a0b23921c3ba2636 : SUCCESS in 10m 26s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/114298bd33a54ef986a7d6cfb90d4f01 : SUCCESS in 11m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0528895ba5f347cdb2783d2b2b99e5d4 : SUCCESS in 9m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5b25093a098c458ab63e6823e0f37555 : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/be8deb154fcc4d28a5ebcc490e4487bd : SUCCESS in 11m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/115bf1bb89c04b65849dcb85149114a9 : SUCCESS in 10m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/23f0cabb90e04e1bb2b63ff3a5fa4c0b : SUCCESS in 13m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70dbccb9859b4587975500f10f3c4afc : RETRY_LIMIT in 3m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/342b29d947e6466fba6fd75bd5e543f7 : RETRY_LIMIT in 3m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc4c949671b141419ce1d4aefba1a171 : SUCCESS in 26m 56s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/23a8d3c66f0e475e8e697f23231382f9 : SUCCESS in 28m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a97b62136400450295d49339b5bb5f27 : SUCCESS in 26m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/614038b58dae4d088d334a80e4bc0d2d : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33113b9b7d474f6f83a3a806629696cb : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b66f9426e24c470c8f39cc02d6bdc4fe : RETRY_LIMIT in 4m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99fe13e2ac5842598cb07ec7b550f314 : SUCCESS in 45m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/592c66333acb44ff92fc8bf430112b96 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7e82d15afaa95f99ebed06fb0486153b24a93fa8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-08 13:58:01.000000000","message":"Patch Set 1: Code-Review+1\n\nlooks like a straightforward cherry pick from master","accounts_in_message":[],"_revision_number":1},{"id":"b0acab8be22ccdd7781d0caf3f89625b37979abf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-25 02:02:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2f82b2a38ef1ed6a8f8c0f8d2b648252e325d0f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:42:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6f2ac4160b4ba9a4440912fea029ec : SUCCESS in 5m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/567d1fa43c8840f5969ab4229cc027d4 : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86fa141e03504b52adce7d3c0c30e147 : SUCCESS in 45m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4b7afc117fbe4b4f86dccd68e0ee5408 : SUCCESS in 9m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41cb445203394ae8b11969ef6a9ff3e1 : SUCCESS in 13m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3334245767724df29f698b8edbb94b57 : SUCCESS in 12m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4844f6c9618d4d978f037405a8c52cc1 : SUCCESS in 13m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7439019c36804c269c5a51cd35a2df52 : SUCCESS in 10m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1396cb7622a149f5b07ca4fa519126d4 : SUCCESS in 10m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/00bfa0ed0a6146b39f401a288bf6634e : SUCCESS in 11m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7a7d3a3582245a7be42c340273c82f4 : SUCCESS in 11m 26s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/bc4e60c545a4414d999cdc00f705ae18 : SUCCESS in 13m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d0532ecaabfc42eeba201ff8bc6067fb : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/44cf08b3f19c45228215dcca7144a66c : SUCCESS in 13m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2fb471e182ec4571a0737b6a587d24d4 : SUCCESS in 11m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a005451051a849b79e7da75aee16614f : SUCCESS in 13m 07s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/298bd02f897b480b96b005844632842e : SUCCESS in 13m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/00375105406547718f7dd95a49f6b7c5 : SUCCESS in 19m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a42b55876a5e4491b083704b36fba381 : SUCCESS in 44m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de0c460bcb804b34be158ec04aadf008 : SUCCESS in 26m 59s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/556effb987d743a7b0286a91ca6ef094 : SUCCESS in 32m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2cbf2256ee944b638d786eb40e37869a : SUCCESS in 35m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6c76d60ad2544a0b09ae6df5570c76b : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81297a06b944a64ae782fda5036156a : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21b41b6362244c8b9bf68b039d1813bd : FAILURE in 10m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f6bd5b7b60b4e978dd75ac573569f9a : SUCCESS in 54m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9e73a59445a845cabc43b28a2679d27b : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":2},{"id":"63f5f328a7706b3e588291f56462f96898d8bbfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 22:41:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1b0694f039d40e6c1c660ef7b9e3057379e933a6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 23:12:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db2854707ff478dd217af2ea596d7da43c8e66d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:46:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2d1d690c7a4e0596bfff5bcc10c3b7 : SUCCESS in 6m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a327dbcdc8c9457bb6dd04f0b6178cca : SUCCESS in 52m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4d39afe5d4f434c90d8862fdb927268 : SUCCESS in 44m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3645e4eafca84ab4910d0eec5848b557 : SUCCESS in 9m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/985f0db10ba245bbb136f312591de9a7 : SUCCESS in 13m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/40e04a9b66344c1d87fad488ad40a3b4 : SUCCESS in 13m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a541962ed02c4227a54922637ac25b00 : SUCCESS in 12m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7dfc002a01e94e8b8b36228ef3b75279 : SUCCESS in 9m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28cf0ed4c3a2447ab0c0b8a37a5dc9e3 : SUCCESS in 9m 55s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b58ea21f553f4a358f782695a7692b63 : SUCCESS in 10m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4eb59b1056e4fd1a105eed9f7b8b7f4 : SUCCESS in 10m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5f272dc433fe4cf1ba5864584f304f74 : SUCCESS in 12m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/975f784b7f2c4628b03429c0274f4966 : SUCCESS in 10m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7580ac87f85a4421897221e7ecad304e : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/85b53dc8e8534569aac5f1c46e75207e : SUCCESS in 10m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/63bf0a4edb1548578036fd68d88e4e90 : SUCCESS in 10m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8044255cfc5b4531a2b9011241478dbf : SUCCESS in 13m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ca181b251e174cb6ba6e7c5a601722e9 : SUCCESS in 20m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e35cad20091d498f86bc43cfa4b35810 : SUCCESS in 36m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53b85bfa38ed4b01bb156e3aa22ddeb0 : SUCCESS in 28m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f1c10c98fd8d4975962facb87d616404 : SUCCESS in 28m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d79189fdaed849b98917d26cd14639a8 : SUCCESS in 28m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3bd8bd4003441eb8cc649a855bfd38e : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34746a70cac452392ef08b4be2e17a8 : SUCCESS in 7m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e187740b0b034d8fa7c536d0192b8930 : FAILURE in 10m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b3af16b1f7146dca11817e6509f6b06 : SUCCESS in 50m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/46330b731da14a538bb5dd30b3e415a0 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":3},{"id":"7e79060a74e463e1cf787c548e7ae183bfbbd066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:31:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1328b2d632620e060ca304d82807fcf252ce896c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 03:34:33.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"f1333e350b5a01b371d2e42c8910ec42e532580d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-01 16:37:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8c0d8e9e3073076dcf207a46a13a415cec7acc06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:35:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46bb4fd19011452c85d94f2dbc671881 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/5f9adbbd55a84dea93d731a33d7b4195 : SUCCESS in 1h 13m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b1b6dcb74544fc7a2c91442cdb1e1de : SUCCESS in 55m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a5236ff9823949f5bb2747798a2aa856 : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e980c08d6edc4712bcc7a00397c66a76 : SUCCESS in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/efbd2084b73348d9a09925de61389b89 : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50c5197d763340a2b1e40f49427dfdcf : SUCCESS in 17m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/74682fef10924d79ab1d271fa713bc15 : SUCCESS in 12m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/717cb3eb0e1044b09e10e53cd1911eeb : SUCCESS in 14m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4af25f1ebe674fdf96a2799282b638ec : SUCCESS in 14m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/520f62ffd21b421eaf4e2ce6e44d378e : SUCCESS in 13m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e96a8e5a18b44f3ba77c0d793f586fdd : SUCCESS in 14m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/385838fdb5d94803946980abc36214a6 : SUCCESS in 11m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f9bc2b9f1623487f98e78987f24da557 : SUCCESS in 12m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a7ab806ac56a49c79817cc89f8f10694 : SUCCESS in 12m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5fc3b76a85764bf19437540d54eb886c : SUCCESS in 11m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/89ac90b3e7354d86935a02e89fb1d530 : SUCCESS in 14m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f891f3856ccd4c46bed461fe1b920c60 : SUCCESS in 19m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3a9c5834682e4676b0f1ecfced83beb1 : SUCCESS in 37m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86c1b0ad3c184959926b544bfea3f49e : SUCCESS in 29m 15s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bf86d8f9ae854e7b96e27476d307b37a : SUCCESS in 29m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc632d737492466fb5c5889fb2e40c87 : SUCCESS in 33m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90a47036b60942e689df086fd59866f3 : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ba74c6f958480791e46c12e0f8da26 : SUCCESS in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0bb097c8bd4c48af976d068fc41fdd4d : FAILURE in 13m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2823c691cc71420ca74c3612442d166e : SUCCESS in 1h 03m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0387eeb955384cf9951d3f1135c2cc58 : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":3},{"id":"d65e9dc6f6b835cd025c6f8574a0394aa90df458","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:36:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"34bc020496351224a90657b24dd58132c6594196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:30:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121c56b2001c4baa8f0bb852d511fb10 : SUCCESS in 5m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/11e5d8f2cb8e46a98cedc0dc113c3677 : SUCCESS in 1h 00m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d58535bd02764cfe8c7fff1499f12c54 : SUCCESS in 50m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/63301a7ef7de4c068c4a8a344112d05d : SUCCESS in 10m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6860bea6cb4d4ffbb0b29b302d7c50ca : SUCCESS in 13m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2618fd7c0a04f90a937dfc97c5fa728 : SUCCESS in 12m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/41fd7c359005489c923ee12638f5cb7b : SUCCESS in 27m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7684ffd7cc0b485db40a8d520c92ecc5 : SUCCESS in 11m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6e44db41e9a4c5a8e63c18e3e5a1d5f : SUCCESS in 17m 20s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5dfb1740645f4f1fa62110e17e562f8f : SUCCESS in 10m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe52e85f53854f61aa1e7d70c3d2786b : SUCCESS in 11m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4b46b079f5f74b9c8ed0219f08df3adb : SUCCESS in 14m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/42ca424ddb004266a41f7e9d1516af76 : SUCCESS in 14m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f367be1fe5274e6ca5437fdf542d7cd6 : SUCCESS in 18m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ae2fc003628f4a2c8e41661c07eb5f62 : SUCCESS in 13m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/67ad9c9a6bf6431fab2234145ddb3c4b : SUCCESS in 18m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/232884d3277b4ad7812ce8ea46c66ad6 : SUCCESS in 42m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9130f0ed90484ca0a96f1815306010c3 : SUCCESS in 28m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/554958445bbe4ae7a4622c3e5549bac7 : SUCCESS in 30m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2cf0aed968514c788161900124579d4e : SUCCESS in 31m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ebd2a7a800e4deaad75ef5d5b83ae16 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28701ab3c6449409adba092ef57cb2f : SUCCESS in 8m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29ad2d11dc5f424891892f384efd8fc0 : SUCCESS in 53m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/07ffba3d895542f59aa4fcaf20c313e1 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":3},{"id":"910f52df1829bcb8c1cafdd4403215dd04d9553c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:30:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bd302c2b101ad8a7970405b4effc169efe3c22e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:32:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0f15a7fb614286be1460435525548d : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b5cbba90f0ae4f90bd834768cc8b2d91 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"96e24bb1dbbac0313cea1f233f8cd918c6d820ac","revisions":{"03769ad4cbcd25adee3e8b7bc659559609de17c7":{"kind":"REWORK","_number":1,"created":"2021-02-06 02:29:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/43/774343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/774343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/774343/1"}}},"commit":{"parents":[{"commit":"cf8f1517eab3a50cdf4744103cc0893c47fdff36","subject":"Populate shrinking shards with shard ranges learnt from root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf8f1517eab3a50cdf4744103cc0893c47fdff36"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-06 16:22:26.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-06 02:23:58.000000000","tz":-360},"subject":"Avoid loops when gathering container listings from shards","message":"Avoid loops when gathering container listings from shards\n\nPreviously the proxy container controller could, in corner cases, get\ninto a loop while building a listing for a sharded container. For\nexample, if a root has a single shard then the proxy will be\nredirected to that shard, but if that shard has shrunk into the root\nthen it will redirect the proxy back to the root, and so on until the\nroot is updated with the shard\u0027s shrunken status.\n\nThere is already a guard to prevent the proxy fetching shard ranges\nagain from the same container that it is *currently* querying for\nlisting parts. That deals with the case when a container fills in gaps\nin its listing shard ranges with a reference to itself. This patch\nextends that guard to prevent the proxy fetching shard ranges again\nfrom any container that has previously been queried for listing parts.\n\nCherry-Picked-From: I7dc793f0ec65236c1278fd93d6b1f17c2db98d7b\nChange-Id: I3a197fecfd68359a325efdfd4590b7bd48598da8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03769ad4cbcd25adee3e8b7bc659559609de17c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03769ad4cbcd25adee3e8b7bc659559609de17c7"}]},"branch":"refs/heads/stable/train"},"8eb552f2ccbab6cfaa24e444ca5b95b7b29f09db":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-25 02:02:32.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/43/774343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/774343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/774343/2"}}},"commit":{"parents":[{"commit":"c250aa74bccb5c122c1881b4c00f8fd8dc2b5308","subject":"Populate shrinking shards with shard ranges learnt from root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c250aa74bccb5c122c1881b4c00f8fd8dc2b5308"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-06 16:22:26.000000000","tz":0},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-25 01:51:54.000000000","tz":-360},"subject":"Avoid loops when gathering container listings from shards","message":"Avoid loops when gathering container listings from shards\n\nPreviously the proxy container controller could, in corner cases, get\ninto a loop while building a listing for a sharded container. For\nexample, if a root has a single shard then the proxy will be\nredirected to that shard, but if that shard has shrunk into the root\nthen it will redirect the proxy back to the root, and so on until the\nroot is updated with the shard\u0027s shrunken status.\n\nThere is already a guard to prevent the proxy fetching shard ranges\nagain from the same container that it is *currently* querying for\nlisting parts. That deals with the case when a container fills in gaps\nin its listing shard ranges with a reference to itself. This patch\nextends that guard to prevent the proxy fetching shard ranges again\nfrom any container that has previously been queried for listing parts.\n\nCherry-Picked-From: I7dc793f0ec65236c1278fd93d6b1f17c2db98d7b\nChange-Id: I3a197fecfd68359a325efdfd4590b7bd48598da8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eb552f2ccbab6cfaa24e444ca5b95b7b29f09db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eb552f2ccbab6cfaa24e444ca5b95b7b29f09db"}]},"branch":"refs/heads/stable/train"},"96e24bb1dbbac0313cea1f233f8cd918c6d820ac":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-26 22:41:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/774343/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/774343/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/774343/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/774343/3"}}},"commit":{"parents":[{"commit":"80b7892266389d04f122dd7b870fd64c98c0fe29","subject":"Populate shrinking shards with shard ranges learnt from root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80b7892266389d04f122dd7b870fd64c98c0fe29"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-06 16:22:26.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-26 22:39:05.000000000","tz":-480},"subject":"Avoid loops when gathering container listings from shards","message":"Avoid loops when gathering container listings from shards\n\nPreviously the proxy container controller could, in corner cases, get\ninto a loop while building a listing for a sharded container. For\nexample, if a root has a single shard then the proxy will be\nredirected to that shard, but if that shard has shrunk into the root\nthen it will redirect the proxy back to the root, and so on until the\nroot is updated with the shard\u0027s shrunken status.\n\nThere is already a guard to prevent the proxy fetching shard ranges\nagain from the same container that it is *currently* querying for\nlisting parts. That deals with the case when a container fills in gaps\nin its listing shard ranges with a reference to itself. This patch\nextends that guard to prevent the proxy fetching shard ranges again\nfrom any container that has previously been queried for listing parts.\n\nCherry-Picked-From: I7dc793f0ec65236c1278fd93d6b1f17c2db98d7b\nChange-Id: I3a197fecfd68359a325efdfd4590b7bd48598da8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e24bb1dbbac0313cea1f233f8cd918c6d820ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e24bb1dbbac0313cea1f233f8cd918c6d820ac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
