)]}'
{"id":"openstack%2Fswift~774456","triplet_id":"openstack%2Fswift~master~I16cb486f0d273192284d217610c476675bd1f202","project":"openstack/swift","branch":"master","topic":"add_shrink_candidates_to_recon_dump_roots","hashtags":[],"change_id":"I16cb486f0d273192284d217610c476675bd1f202","subject":"more test assertions for shrinking recon dump","status":"ABANDONED","created":"2021-02-08 13:40:44.000000000","updated":"2021-02-11 01:05:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c68495875f461e79eddbe0d27d1d149f2198afc7","_number":774456,"virtual_id_number":774456,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-10 20:54:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 15:24:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99b9956bb9499c33ccd1267b84cb962c1f7114d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-08 13:40:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2642c34a37b448d8f2efbd50c5a572fea33d9e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-08 13:44:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bffbba47b6ec9ed4abe7041136cfd06ea1895e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:24:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b22d1f2dd4504e47a121cf22b0e32a37 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/1812b68f6e9647b4aa298b6d58957c21 : SUCCESS in 58m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2b939b120b8b4ea99ef688bb4dcd2c02 : SUCCESS in 42m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe3c427810ed43fc809eca679039f085 : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5f83009f3e0a4d4bacb0ecb4f6c4a1b5 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0160f0d55e14b3d9b1385c408ac3fe1 : SUCCESS in 14m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff2ca0e1ce1747289b8c6d9d10bb0c81 : SUCCESS in 14m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/312727e8946440ee80eb62915ef44da5 : SUCCESS in 14m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43546550a9a74d5b9d6f5af9207f5b5c : SUCCESS in 14m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e06e0e1e38fd4df8bf76c950a9147ad1 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c2bae99eae3f4a928c6121738170905d : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0e91470d28d14f36941e96fefcc5a3a6 : SUCCESS in 18m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0f1ba95b58f947d590026530ce20efb4 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7331dc67e98d4b869f3e6eb8afb90158 : SUCCESS in 20m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd4d9308f28a46819e5fe200a4701107 : SUCCESS in 17m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b699458fe464fbc8ba9f6f6e6b089a2 : SUCCESS in 45m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0ae77de4320c4f69bb99b05fe98a2bec : SUCCESS in 51m 22s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/55a0b905a0f44af78637972d72fb628b : SUCCESS in 47m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78729f046ed04afd9a7e362c96af7495 : SUCCESS in 37m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1edae3cdbcb647b6a234471536fb0576 : SUCCESS in 36m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e0cd9982f554a1f9ed05bab99ce6dfc : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70da7b63cd9a473a937b3e1f474b3d57 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f28fa2c182f14f1cb3a7874ae167ce1d : FAILURE in 9m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1737a16a6ca3429a97fb15d8b3249fdc : FAILURE in 1h 13m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b21bbf71e6d5dab2fbf8f3a4f593d43e3de367c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-10 18:51:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f96023f5d2db1aacf45123bc5a6adc2a04b3bc10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 20:54:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abdd5523fc8a4cdf83e1fdeb12fb36cf : SUCCESS in 10m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f9ff7ece6414b9f82b725327517d7d1 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/afa9d9cdadbc473c878cc021cdd3814a : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19cf016998144a3aa6cbfc9dcc439b01 : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4099b3bf89d04f99ab681f4b34818d34 : SUCCESS in 15m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cac6d62ec0c04b928793989471245d56 : SUCCESS in 15m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3efed6dc7e8b4640a34f0ca40a0aab15 : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1b21e15d2b44f5a819299c3654aac94 : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21d3796e6d0c4baf9b149be7872dd640 : SUCCESS in 28m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1dc28208dbc348e5a1406f3e19acbe90 : SUCCESS in 18m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c94b95fc20774117a3f7d6f452ac563b : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3be915cd87d947d9a624e0907f92c4d3 : SUCCESS in 20m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ff5feac9203c43b2b12148a68cdfd7c3 : SUCCESS in 20m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e63f56efb4a4e3b8a968d32b933e165 : SUCCESS in 42m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b0075d2c20a473eb1ce9eb107b86636 : SUCCESS in 43m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d5b0c0ba6d45abad8fd10112b0ffe6 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2701ac90d5ce49a09df5a844da6ff71a : SUCCESS in 8m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b48fe4d9ceb340409bbad6d476ed138d : FAILURE in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c68495875f461e79eddbe0d27d1d149f2198afc7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-11 01:05:10.000000000","message":"Abandoned\n\nMerged into parent (Add shrink candidates to recon dump)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"418c227f0fbb270cee089cff3f8bf2a6619bb5ed","revisions":{"b101b03dfcea9c7bd5a6a0af76144bea295a6c1c":{"kind":"REWORK","_number":1,"created":"2021-02-08 13:40:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/774456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/774456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/774456/1"}}},"commit":{"parents":[{"commit":"b9337030f2734005fbdbe8daa260b71b973dbc9c","subject":"Add shrink candidates to recon dump","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9337030f2734005fbdbe8daa260b71b973dbc9c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-08 13:40:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-08 13:40:05.000000000","tz":0},"subject":"_zero_stats and test fixes for shrinking recon dump","message":"_zero_stats and test fixes for shrinking recon dump\n\nChange-Id: I16cb486f0d273192284d217610c476675bd1f202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b101b03dfcea9c7bd5a6a0af76144bea295a6c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b101b03dfcea9c7bd5a6a0af76144bea295a6c1c"}]},"branch":"refs/heads/master"},"6fc7eb8bacf7f58f04f7e6304c84df5c528c4696":{"kind":"REWORK","_number":2,"created":"2021-02-08 13:44:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/774456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/774456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/774456/2"}}},"commit":{"parents":[{"commit":"b9337030f2734005fbdbe8daa260b71b973dbc9c","subject":"Add shrink candidates to recon dump","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9337030f2734005fbdbe8daa260b71b973dbc9c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-08 13:40:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-08 13:44:05.000000000","tz":0},"subject":"_zero_stats and test fixes for shrinking recon dump","message":"_zero_stats and test fixes for shrinking recon dump\n\nChange-Id: I16cb486f0d273192284d217610c476675bd1f202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fc7eb8bacf7f58f04f7e6304c84df5c528c4696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fc7eb8bacf7f58f04f7e6304c84df5c528c4696"}]},"branch":"refs/heads/master"},"418c227f0fbb270cee089cff3f8bf2a6619bb5ed":{"kind":"REWORK","_number":3,"created":"2021-02-10 18:51:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/774456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/774456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/774456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/774456/3"}}},"commit":{"parents":[{"commit":"44295adc2ce8c123fde59c90f716007159caa0c6","subject":"Add shrink candidates to recon dump","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44295adc2ce8c123fde59c90f716007159caa0c6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-08 13:40:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-10 18:50:53.000000000","tz":0},"subject":"more test assertions for shrinking recon dump","message":"more test assertions for shrinking recon dump\n\nChange-Id: I16cb486f0d273192284d217610c476675bd1f202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/418c227f0fbb270cee089cff3f8bf2a6619bb5ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/418c227f0fbb270cee089cff3f8bf2a6619bb5ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
