)]}'
{"id":"openstack%2Fswift~774481","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I952690558486a1d981985eae0fe9d7e07ff30e17","project":"openstack/swift","branch":"stable/victoria","hashtags":[],"change_id":"I952690558486a1d981985eae0fe9d7e07ff30e17","subject":"Prevent upgrading to pip 21+","status":"MERGED","created":"2021-02-08 15:37:41.000000000","updated":"2021-02-12 07:45:05.000000000","submitted":"2021-02-12 07:43:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774481","cherry_pick_of_change":772424,"cherry_pick_of_patch_set":1,"meta_rev_id":"a38e3735a5a72794d783b17899baec53e69ec032","_number":774481,"virtual_id_number":774481,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 07:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-08 16:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-08 16:50:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 16:47:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32326168727e1093bfb3583c3c206763dd170769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-08 15:37:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"83bf5a18a7c2f74cf84ca7674e8bda56feb10d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:47:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9c6a67d41e4d518f45b8a9d5766f3d : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/66834caf0ea042358b2b9f0036408b1d : SUCCESS in 1h 07m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/690652da2ae74355a06748d6b9df03df : SUCCESS in 56m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e228aa39977451c905d631f4526f398 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ed42233f92314c368ba88e84dbb6e492 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4bee607f64bd412da7913df16e7e3a5c : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a7adb727eb34669832986f2abe997cf : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5a0d8ea107bb406e8dbfac86adafef92 : SUCCESS in 13m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/358f50a13b3840c280d69159c0a39891 : SUCCESS in 12m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e561f2d9f8244cdab110b977f6b0adca : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b4eeef8cdf34d81a1e39a370d6d8463 : SUCCESS in 29m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/07c3cd3ee02c4f89956654f0c983cdfd : SUCCESS in 34m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f4cbb901d84840dabd3b061d2ee4fde2 : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/edc3aa06f968424396a618ccedfe59bf : SUCCESS in 43m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8e4047640a4a4febb8e0a56c2607b5e9 : SUCCESS in 18m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/465bd427031f45d49c3cf4e92fbd6b01 : POST_FAILURE in 5m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a6897a6f7e4e4ce7a1518a1252d9068e : SUCCESS in 48m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eced048dcb9d4ce9b195341f4a6d207f : SUCCESS in 44m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba6973fca2dc46569b58fccbf34d8773 : SUCCESS in 44m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78ce025aec0a473c9f590e1f5b74b6d7 : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb7d917ed9b84880819645888c8822d3 : SUCCESS in 7m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2a96188f1064a879114c7d0631eeb54 : FAILURE in 8m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bdb571165b84c5d85536082d4a270d5 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1dd5ef5098793145957168fa8dcf0529af717655","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-08 16:50:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa9d5119013946ec8250d39ac514f6afef31dfae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:51:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"901243fcbe53e7fb581a1857a8972c9ab5728cc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:59:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70494b94d3bf4efdac4e5befa94a9ed3 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/1e3a3e79aad9496ba7a7d3345ac27533 : SUCCESS in 57m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef10d73bceb242248b8b949f0984b1e3 : SUCCESS in 44m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c1912df993484dc8ab44dc934e3acba1 : SUCCESS in 9m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ecad813574cb4655a674b6be81d60b10 : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b34f1d81d104a449dd6d5af64b193a4 : SUCCESS in 12m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/818b304f99ca47999f08eb0a6ea9b099 : SUCCESS in 15m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32e122029ac74baba7978d5db48e4677 : SUCCESS in 13m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/093264280ad3421fb55955b95c3801f2 : SUCCESS in 12m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f49e9180345544528eb718c5516da3f8 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/352ef679f0b7499ebcd14e9dacaebd6c : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d0b1b1baccd4d6f8896102a0415fe7e : SUCCESS in 17m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/05b18c2fafdf4bb18afbebe87835cf36 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a67def7166894868997c963ab9371e32 : SUCCESS in 17m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b92659e900a64d7ba2443e1dc0b2b1d7 : FAILURE in 42m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd0f5aad704846d58dd0b9734c0f6cfa : SUCCESS in 37m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a23ed6a30c814ed1a45c5270bd5452f1 : SUCCESS in 41m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdf80e3c4f6b44ed9b3c65235f84d744 : SUCCESS in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9139abbb38b4391b9918c3e004fcda0 : SUCCESS in 6m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff1fd676af84fa59e88aec4d35acbc0 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":1},{"id":"16a568255cea516d2ad0ff4674e0342b3fafa7b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-08 20:59:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"764bad7d442e2d772ac01024b5b77baf893305bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 22:08:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3a409d503c41cb98fc2b5dcb9b0692 : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a2406db308894d8baa2dd1be9df69cfe : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0dcedda50bb4dcbbf2a9bc3ae4cd055 : SUCCESS in 56m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73ebeb9347f44bc5b40618138b1ed407 : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e3b4436ccfea4132b1e7a51e5f2af2a5 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a261bc213b240a2a23fad2033dae7c6 : SUCCESS in 31m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a59c097dbd644d49ac8bb1395d4d9d3 : SUCCESS in 18m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d97a29eb1daa4bfc84b5c4cdd42f6584 : SUCCESS in 12m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd22f97deca438c97fe6a4d90ba67d7 : SUCCESS in 11m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4ff24422c3f44a2b6220de35e18d9d8 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13056302d71d4382a66de2d478851b1b : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/08a384a5bab448a0821d051f556de2c8 : SUCCESS in 15m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/379eb96381134a0188a2220ad71101fa : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b5533b7a9ae14397b3c80ffa809c5aa7 : SUCCESS in 24m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a10dd2eefd76462985c0c977a4fd94f9 : SUCCESS in 26m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45068076194c4ad7b8573d59e5173ea9 : POST_FAILURE in 5m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/525cd6e947a94fbca177ab6b86542ea8 : FAILURE in 45m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8859478e5cc048a0944efd5a8829876f : SUCCESS in 45m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13316075fa0044e6bf8eab416ae6820b : SUCCESS in 36m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73f9b21a438f40e99b371205cc03727c : FAILURE in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6aac3c1694545a597a6a4ecd411e906 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/819b72aca80441df9ecc0eafd18afff6 : FAILURE in 7m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/356f998611354922b389a95eabf0ca41 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":1},{"id":"04a51193a52e9c7d41db77edb3436919aa6ee061","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-09 18:01:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05e9f5e037549ee4912bc1af6f45f20e1faa8b7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 23:16:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb635a6b00042bba13997ff52344ad4 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d51437e436cf4d4fa7d2f62ebb1974a3 : SUCCESS in 1h 05m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4dc55bfe2c0847c78957680821d8fc10 : SUCCESS in 45m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3189f37c3a8546d7869d5c876e5e57a3 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/215b07be258f4e89a12b4c2b6ce5bcf6 : SUCCESS in 7m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06b8ccc1aecc4b43bc7612ea05232ebb : SUCCESS in 11m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e8722e3dad47ff85c191400ea04c82 : SUCCESS in 11m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50d0db778f154c548501e69f6f2a8248 : SUCCESS in 12m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3e47569d24a04b839c938796525d724f : SUCCESS in 13m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4edbf0506184476f99b8e92034cc3d74 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f108d44b90bd4f47b42c52c70a596940 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/75b7e7d782f64048805a8c34960ad392 : SUCCESS in 16m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/08801458fe6d4fdf8f86277003e50258 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f24ece6db1c84f868fe68e231da43062 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/265c67f3955f456190d7c0ef94ff9a18 : FAILURE in 19m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c84f53795307415b9d1960fa835e3d22 : POST_FAILURE in 4m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c47f95a9df4844758f82360b29b8bd40 : SUCCESS in 43m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62c959d0dd9c4a86b93f3cec98f4a4c7 : SUCCESS in 41m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32fb69979d1f47de96ee5b64d56f84ea : SUCCESS in 49m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58b78cf65507455ca4bbc9d2013e2443 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fed039ef2ab4ce4802d992724520f41 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/024510f067844022b535d2fd4ce9b562 : FAILURE in 8m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5162c6cd685442348905878b6a479e26 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"208e22f38cec0e8ddef8191272b18dfe06367d77","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-10 00:06:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3f8eb062d1d2518bc9fafecec306669cb3502b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 02:14:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512de9737a694bf883f72aa8e7214466 : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/4654335be55c41818693babeceec76a6 : SUCCESS in 58m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a87f8a3cb35e4603ac56bd016a09b8f4 : SUCCESS in 42m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8c8cea77e0b4b7fa6a898aa26363ffc : SUCCESS in 10m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f10df3fe73774a5281a055557549aa30 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2600122b013c450ab19166bf5dacfcbe : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0122ec4a7c87454390b66c049e112434 : SUCCESS in 12m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6632625ac204475987d092ed3249dc12 : SUCCESS in 12m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/12bce288b3dd4b128cb3814aafb89190 : SUCCESS in 12m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5ce7cb2f22cb4f9ea9bdcf4dd090d007 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25d1752b61a34b75964313342049d861 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1222962d6bec459484da1ac07d48fb84 : SUCCESS in 24m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/71d0858759e3480aa9e3ebca6d48b88f : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c6e854427a264837b39fa6258d722016 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e88d4d336e1f49518292235f24faf0ce : FAILURE in 19m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2a215c329944606a6b9193b17cfa4e6 : POST_FAILURE in 6m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ff1f5fc03ae49708c43bd0728e73f52 : SUCCESS in 42m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e4c41bceab44bdb8837380e03624d98 : SUCCESS in 40m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bba6faa974f54d5583c403721f767d40 : SUCCESS in 36m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dee3b125feba4442a9f99dadea3d0d3c : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671ca56e98664735ac6eebd0b56df611 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a46a7799041451aaf9dc60a5bdf81cb : FAILURE in 8m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75888c9f12494af29451a407edb56d00 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0920ed4454f89a3bb7d5290ef810d192b9d40319","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-10 04:59:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2eca6df64e7d121c57a2357d565a178bee30c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 05:59:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e1b3fd563e4bebb2dffefd28b9b248 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/c4a7b6576adf459bbfad2b35168ab58e : SUCCESS in 52m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0a202d89c2e4e769431602bfa049044 : SUCCESS in 55m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79f6f5494918429485a4ab68d7cb2667 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b62458366e454125afe9fd4bb4591154 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b9d7ad3034f4a719851b20ea54cb9b0 : SUCCESS in 12m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/baf48a9bf87a4959900167de90d9db3d : SUCCESS in 18m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4fdfe8ccd079434186bc62d7aea198d4 : SUCCESS in 19m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fbadbbfb4b3c4a0a9dec059efae283fd : SUCCESS in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d94c407396e405cb13cad2b112c2965 : FAILURE in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a6a94c72c6524e4695b512b5a020a39f : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5e37af13173e4d5fa994aeb39f389197 : SUCCESS in 17m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d1b2515e4b674abaa168587f0343ea01 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e203a1ee2cc04937ac1a0ac429484808 : SUCCESS in 25m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9ad7dc11431f475b8806925d9af86fbf : SUCCESS in 16m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2651780414e049c2b54df898eee04bd5 : POST_FAILURE in 5m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/99726e53f1f4415a9991031f39167ff3 : FAILURE in 41m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac7c4be462a84acea55cf3ccb9ae9ecc : SUCCESS in 36m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78ff5c9c91e24c4c944da56b14edbea1 : SUCCESS in 40m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9b7286d17be4a328d4d79ba2eb15c7e : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd45b6273657436f992d03b5c9888348 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0673e2bcfcf14656ba8b91c7e0a41d30 : FAILURE in 6m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80fdee9990cd4a70bb18906d9be5eabb : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5909d7ce8ceec402578fd6157bd7d79f4d6cf01d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-10 16:09:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd1f3d76b4f6e2eff77088daa96636aac95a86d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 19:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc997f542364602a828156960412825 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/35b9457f355e4ff48678a67d69f4a153 : SUCCESS in 44m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca6eb241161f4c89b7349bcb87710891 : SUCCESS in 45m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1912fa274c1d4a8e8a4f62325ca28b47 : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/799f62f805314f7ca6499b65eec86b94 : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e08aceeb00f141a3bd1de83e98e3327a : SUCCESS in 12m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8d56f4fbe594a7dac24dc72477e2a45 : SUCCESS in 13m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/543370518358478ba7ed482a4fd6a8c1 : SUCCESS in 12m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ba7580df5bad4639b6d7fe850dc9cde7 : SUCCESS in 13m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33be0122790e432aa755c0b6e4a37a02 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2af524fade1345d98ec7b8a635aae593 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/becafb7640c44ddc9a844122daccb552 : SUCCESS in 17m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c54590894c8e490d8f8f7b4f28c1e1ed : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0b4b5db8a89c417d8b14e02fdbedbeb9 : SUCCESS in 20m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ea5a8b210ca24b628ad6f65d475928d3 : SUCCESS in 18m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3dae4c1e6dba45ebb90a6fb320e56557 : POST_FAILURE in 5m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3618d0cb2c544240b954d5b90d67b6f0 : SUCCESS in 42m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f71747923c044eaa74683f8823082ea : SUCCESS in 54m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dad9a933ea98439eb3378464a77dbb51 : SUCCESS in 38m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac1b4ae1217b42628a018e5c4de084f8 : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2530066e44a4217959c1a2c1f932b16 : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bdff4560e73b4c7abe1222b4aeb8333a : FAILURE in 10m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a2813dab13040beb0818bac64ff4c25 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a41b775ae0b5722d7f55eb036ddad836368f82b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 19:07:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05e852480c713b18180897b43dc7ff7af950029c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 22:13:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46121b8aa2eb4becaffd30587d93d230 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/876b5bf7c6a54e78b2bdcede252fd407 : SUCCESS in 1h 10m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/72726a26124849dfa516adaa11056a47 : SUCCESS in 49m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cd5666c6515d4d86af5cf31d3ee66649 : SUCCESS in 12m 33s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/91ac45f09563411b92f9c4f2a6c4d857 : SUCCESS in 12m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89edbf8b01a94b70bb8005e7b9e846b8 : SUCCESS in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a84e55cf25d4dde8ce17a68d020cc67 : SUCCESS in 17m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1044703a6e0c49468b350e01fc0134bc : SUCCESS in 17m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e99303ea4dc4a6db15f0a3c547bd983 : SUCCESS in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/02808b5f5d014b2ea982228bd03c14b5 : SUCCESS in 21m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0e73da0c24f44beb18fedb517810a01 : SUCCESS in 23m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/467f5cdaf500475b89dff16c5cb2b16d : SUCCESS in 17m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45b828da83c4476daa1a0b57e57ccf08 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ab34ac86bea045a886d5dbcb2f94de37 : SUCCESS in 19m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/01ef34bd81184ba29ddfdc867ef9e835 : FAILURE in 1h 10m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/27c890d1878e42b885e42d737242539a : SUCCESS in 1h 04m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cefd292cc0cd4ee18334ea98d9738889 : SUCCESS in 1h 03m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2331456faa664078bde44b8f93a8bae1 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2e541dd5944daaa58aa85da11e1e24 : SUCCESS in 7m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f86c6f45cd14c83935b526590201253 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ebd50d51aa364544fa97967b63559dc3fd51aa1a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-10 22:49:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e92558904040130984865230f28b7ef376437d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 23:58:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a82b54213cc4696a02267af1295f29f : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/89e520fda7804076aab46db2d96f4338 : SUCCESS in 1h 06m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/afacbe7a4d3d45e08c1cc5d9dd8d7fb2 : SUCCESS in 49m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f15ab043024b402c922fc39164db185c : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e0618d21ed7941f09d300c9e0d3937fb : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0d14e6849e44f53a8b533e4a1074870 : SUCCESS in 12m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0fac4c19153448c79985929e0af9f1e3 : SUCCESS in 12m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a58f80cc20f8446fb4bba70fe52da212 : SUCCESS in 12m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f0c1f40014414698477f9bfcfb5e52 : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/345a0609d5cf496db74b560e2fa797b2 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6f31c2a85eb54f3db91f5e5dd2794c11 : SUCCESS in 24m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d1f87d5f34e74f8184a7a6506047ecee : SUCCESS in 18m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2b4be9bd75fa441b9642891f9c888d15 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/88a8e09e91414a7393bb0a5965224605 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/281412d97f7e47869a09e97a49ede4a5 : SUCCESS in 18m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/67866ca2f57f43828384ef06007839b9 : POST_FAILURE in 6m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cfb736bfaa47437eb66dfe6ba1c57fa5 : FAILURE in 43m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9319112dc53145bf9d9abd5a208af5c9 : SUCCESS in 45m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d68d7b1fc3154f07bbe6dca9133b15ee : SUCCESS in 44m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6a895778e9d4e9193969582fda4bff9 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109736c4e2b846af90256c7a40cbb132 : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68f43278201645bd93004f6dd1061fa6 : FAILURE in 7m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe55515e0ce74bae9a042e76f0a32dbc : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a5e632794b128543d6d7eea783e3921ea7adf974","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 03:10:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b2646d466118499e459a9152e34a0ead71a1d3a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 04:15:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad5854c938a40e8b09ed1c426d9d0e5 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/7b862e8a3419443ca2bb63dfbd5ca9d7 : SUCCESS in 45m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26c35afc65d445f0ba4b474bfcefd1ca : SUCCESS in 1h 00m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2649fa859cf46fea03c87aadc5102d7 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9f7ef285e6e8454cb81e5c4d1c40060c : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca43fd1d884f43c2a35e7a4aa4fb00df : SUCCESS in 22m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f46cab0278b74b20a8b3936b5b07f3d8 : SUCCESS in 27m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/420526ef897c447a904a79de826d54aa : SUCCESS in 12m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8505578f886847458db6f155f9a883a7 : SUCCESS in 27m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f11d27f71dc4e469ab8bd8430101a92 : SUCCESS in 30m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e36a5bd09b6497687d20bfb12458bf1 : SUCCESS in 49m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d32aed35daa94e03ba7975befe57529a : SUCCESS in 26m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/52c2e0a00a8a416da4409be595b954a1 : SUCCESS in 49m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/74596d286c98450993380b05f841f154 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/80c294221d714063b6205e6527ccd577 : SUCCESS in 17m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c7db5ba95a0941198f8f35ffcd70fc51 : POST_FAILURE in 4m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/754499252bdf4f8894adab331f0f36d8 : SUCCESS in 43m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/820be20007604b12b415fc3353839ead : TIMED_OUT in 1h 02m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e0464de70374e4ab0077fc8e25ee649 : SUCCESS in 38m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e6cbd6e42a64a5e9c0b19996e57d38e : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ce7d35f4d746cd9fb17f1d51c1e913 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc4f633f53784d46b723e1f7382b9036 : FAILURE in 9m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e3de2be2cfa4592ae80ce179ed8c782 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"360d772ebae4df1119adbb0acbe45e304cc2a31b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 04:56:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6aea93d903a5b116f1518eb6e47ba43be8220b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 05:55:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075c30436b674e8fb1fbcba2a6f64ad9 : SUCCESS in 6m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/5fd2f4756f3f440eb73decf16f82aa7b : SUCCESS in 48m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0d2dc2c905f43279755ac7cd48f542f : SUCCESS in 39m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34b9174610cf4cfc9c14b899c3d0e7e6 : SUCCESS in 11m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6aecdb667e354519991df994de63152e : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6998e1706fcd46248d237f026e95f054 : SUCCESS in 11m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/649a66cf1641437abf8b5a94f4cb7704 : SUCCESS in 11m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a1f3291a33f49318dcc465a49f1d83f : SUCCESS in 11m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/398984b853144e79889d939e2c321257 : SUCCESS in 11m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/719d4fc3ec7e42938d0258b73ef92d25 : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa95a7785c304d8e960fccb9fd15a3f3 : SUCCESS in 20m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/59b20aad5a7b41b3a3cb24385af73867 : SUCCESS in 35m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/293f4c97e4274fd6836e0c3470a1c8bd : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d2eca39be23b49239a7a7aab446ac5e8 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e7acf78d303f4e9ca8951c8a26b7697f : SUCCESS in 16m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10c1550a942b4f4697b651298ba4e1e1 : POST_FAILURE in 5m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/48eb16ea50bc4855a024ad3344c50edd : SUCCESS in 38m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48a942959c964dab91bb2b573c5a55ef : SUCCESS in 36m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b5e2a74f7d0456bb24ef20f0b65a572 : SUCCESS in 45m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afdcd7e97cea442a9aacfc2e07c4227a : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca29781f5f54ae9bfd08bde8dfe2954 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/95f576148faf49e78dbbe021f8176d8a : FAILURE in 7m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/362dfc29f3054cce975d725bd3a75e28 : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d3a691cf18d3c87530463164b0818824dfe36f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 05:55:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97b6cdbe56fe6f02b753c9fa09508f27eb7ce9b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 08:41:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf605408c2e4d96ba01c7edb856dee0 : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/bff494ccfc6444ac99721581b15a68b8 : SUCCESS in 47m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db4b2b6ea2d44140ace1332ed09b17f3 : SUCCESS in 48m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6701e8a4237242ec837d1061294d031c : SUCCESS in 12m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/971214f68b0949b2bfa101123e951aad : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83247e55c26a46229d739442c6823f2d : SUCCESS in 16m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b27d42497bdf4734a4f174789607fc25 : SUCCESS in 14m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5fc07f38c7f54ec4a21dd346626ac680 : SUCCESS in 19m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03892a2069a74e1191cd2b9e779471d3 : SUCCESS in 14m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5dcb0d4eea5949c6bbf57487fe74f1d0 : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0141c88daac841aea89f8f4fa8550ed7 : FAILURE in 21m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d4c06dc80b714a97889cd6c46bfb48fa : FAILURE in 20m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d7b97b56300945aca2cb7d8665fdd5b0 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/abcb80405c504e7ba1d28ad2ababa549 : FAILURE in 20m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4e46d78f2de0464e875e3644adc0fb4f : FAILURE in 41m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c211af16ea924751a6ae27533bf68283 : SUCCESS in 46m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2cc0abc56900450c95f4545280a66523 : SUCCESS in 38m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8376fa81023443fc9ccf478668dfbeb9 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80ab3e8460e4b85830452b1302175c1 : SUCCESS in 7m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af9ed892de9a4657b4519d96a3acf1a0 : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e12ae87869a29eaaf3ff1a0c7ddc9a19a7f86f27","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 17:18:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a50f94687ad145873105fe6872a4b61c8fc10988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 22:27:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc0b86e0e564c55aaa361e41ca328d6 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/44b0f37bdc2b46f0aae451a17e5d2642 : SUCCESS in 59m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3777e89027549a5923a0552f710bbe4 : SUCCESS in 42m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2114b235316495794b807c620a9c526 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/52d3691fdb924e3193e1f0e83d104ba0 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/484109bebc164b9293c297f184687b09 : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b45ab0dde5334eb69ccfc74f6ecf38f4 : SUCCESS in 13m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c25faac85bd49b1ad262dec69e4254a : SUCCESS in 13m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9712847cbe2a4513b6f8a37f63299859 : SUCCESS in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1855123df6844765af4054c112b819bb : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5e74f70594748f6a62df728dd913b24 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/296fb3544c53482bada4a98bf88bf166 : SUCCESS in 17m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cde32e6dcb404520aa6d7955fb6508ad : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0b6f35d80077442dbebb6b9f3d4be863 : SUCCESS in 23m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6b347414008644d5b2b6cbb5c4dfd415 : SUCCESS in 19m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/636a29aa2d0649fcbbf4d31f51bff77f : POST_FAILURE in 9m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3fb471a8f2d7407baf8dc85baecac7b9 : FAILURE in 1h 29m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9fd979092d84ce789bfab1c2648bdfb : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/28a40e230a884be39b18f38435dca9bb : SUCCESS in 43m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2134e2549b7b4f4780f4959f1cc50dd6 : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b37f5bf496460d9fd9d83abb5bd3ff : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88d27e8d951442efa270ea00748c2237 : FAILURE in 8m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e23023f102e34c61b63738f8b5779b6e : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e23351a3dfe6d7a0d5f099eab3f2bc1f9bb8e965","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 23:21:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"80c134454aacf3648b46d9d842f6daae596308e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:42:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb18de181f64b52a52fb007df5d5ab4 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/4d81fe3034ea4e298edaa92bf46dd829 : SUCCESS in 1h 01m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6af186b3eeaf47cfb31de4550952c0a1 : SUCCESS in 47m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b425f3145c8c455a808fc314666a357e : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1a45668b612d43cbb34146119515188a : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef6c5e14faa94f1eb5fe2f8b8ab23a2b : SUCCESS in 13m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cbca4632120048a1a8de3af2446a9cd6 : SUCCESS in 30m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/99848e53f1284054877969c7364b0e4b : SUCCESS in 13m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e9e901e47e9d483c8f2ba053f44aa0cc : SUCCESS in 16m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bede2e5db6b445cf96b652f05a0f5451 : SUCCESS in 21m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94495c917b1944f2a441a3673a7f91d2 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60b3a870fcfb4a08bfffd12bb947f203 : SUCCESS in 34m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6f173f1797ca4b418532715c4ba31cf0 : FAILURE in 21m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2eca3822dbf6445880d586b1c8ee278f : SUCCESS in 18m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c0e00e6811c4b86ad5a40f275c5d558 : SUCCESS in 19m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd995a37df47465ea23f586c2098c3fc : POST_FAILURE in 5m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/84cfca6ff7094e11bbb70bf5f89f5830 : SUCCESS in 46m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/65124174023c45218b5dde3650dc821f : SUCCESS in 36m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c867baa8729f45bdb5bf60d93e2ad8fc : SUCCESS in 41m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b54ba922fa7b4e8cad6183208e4059a9 : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c33ae078114c98954870e39c0b4d6f : SUCCESS in 7m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8b0434c624f45888071644da06268a6 : FAILURE in 7m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74c4fd3e921b4a59a01907e2b59812a1 : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d7e012934f8bf87219174a881892b7ebbc3e8efb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 03:06:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f655ac6fa075b3bf08af9aaf033c3b5f2a71649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 04:17:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0758c3db899429bad4a43d97cc7675b : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/9a1485fe166d4566abd0c4b19d66528d : SUCCESS in 1h 02m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/250a36b550604ae2b64a6d2a369f2b30 : SUCCESS in 48m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24753f5d7bb14152a70bff809df75b2c : RETRY_LIMIT in 1m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b8ed4f698c1c4734bb8be1ae3ec0cce5 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/860eefdd41ef4789b0f304e1dc7c37b3 : SUCCESS in 13m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6ecaf32f1b647c29721592697003f77 : SUCCESS in 14m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/81f86da6941c49069093b256de88568b : SUCCESS in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/007c09e81ac34fe889630e717b918a6e : SUCCESS in 14m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc942b6c666e4b0897205b0252fefa7c : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ef910e388194de282f2fbec2db2e5fa : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4df4225add344ddda319a8637f522db9 : SUCCESS in 17m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7f7d10f2cf0147cd9ba55d45d020a639 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c98c30c4f0af4ce995d41abc1d1ca371 : SUCCESS in 19m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bd6befc8d5a34df09321e4bc27f403ba : SUCCESS in 17m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92680b485dd74acba380c204f8cc3aed : POST_FAILURE in 4m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6f4da4ce72e41549d13f96fc26879da : FAILURE in 43m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4cfcea12f4694381a3a124a8f074b87a : SUCCESS in 41m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a544c652d5184fe9a6c0b4545439cdf2 : SUCCESS in 39m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5337b310a7c44149b6a1cb33ae9bc968 : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c958e7db41f84ec9b8d2ed4ec9ef7d46 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ffc637a572104729859d9445d22fabf4 : FAILURE in 7m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9f240d6558749ae8fd8114bbdfc6ce0 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2051f7c5a046d15c36171085420d80bf9e6da368","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 05:14:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4dc01794a88c756004e0b4280f0d50d67f2048a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 06:29:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e5b8cab6a84b869343ead0d2449f4b : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/9be0e9ed464f477982a421536ae21552 : SUCCESS in 1h 09m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/213ea2570a6e486dbff347071711541f : SUCCESS in 48m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3af7b9d3c1740a4bfe7ae8920bbd9a9 : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5807fc519a4b47f5ab984a6a40b8d086 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d746804b7e34d0884788ad8a16d85ca : SUCCESS in 21m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4e44be7e743f43268ff12c7bd94b2fc3 : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a95287a2ca04446e91d99946bfc63879 : SUCCESS in 14m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c6b127be47174516a98565e9d3127be5 : SUCCESS in 12m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d89eb40fa1874815b8b7d7887df55a2d : SUCCESS in 24m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb5ac66ebeb5419bae484b685c558deb : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8fddfaa4d2bb455993509786f5fa2fee : SUCCESS in 17m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d4e0891626af49ada676ff12fda06fea : SUCCESS in 21m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f7ac7b8291fd4c7d921a2c68fecba0f4 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d87eca634a0348a8b07500028e708cee : SUCCESS in 17m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d3e238c81d14d63ad689c89a09b921e : POST_FAILURE in 6m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7078e7e1ff534ae59224576b2188f28d : SUCCESS in 45m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/050af8152ff344ef9438f1245cbec5ca : SUCCESS in 33m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/632b3ecf607f460c8727869cd1dffb4e : SUCCESS in 45m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c277c152cce4a9e92b13239d3683c9c : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1485335454404196a52572de7267205b : SUCCESS in 7m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/774927a82c014beba9bdd151219d33c9 : FAILURE in 10m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7520701801e4ec9b630798196605d98 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":1},{"id":"78a2496c7f173fd0a726c5e868ecd452b67f825d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 06:29:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1570f40628d7b59970803402c275d65602a4dc0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 07:43:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41212de25abf42fb89bf931f222052b5 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/003cdaf882054a4892f85a582dc484c2 : SUCCESS in 1h 08m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5967171fbc6c421e8345cbabcbd97e09 : SUCCESS in 45m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/78eaf1d5da0b46e08324c89094286f32 : SUCCESS in 9m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4fcb24ff2edf41ac99d9cfc178f81190 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e058d7d4769b42d3b06104c720d1f801 : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59ade5bcf25b461e8f4ae26bd22536be : SUCCESS in 12m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ffcef3ef915c42c4865e5c42e18dd54d : SUCCESS in 13m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e568cd356dcb416293530ef6b1565675 : SUCCESS in 15m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb92e5026f54402baed11d5d9fa69fa4 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee8c04ca8ea540d39a6bc5a046dbfa83 : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/994c141cbf464da58cf6c4d59d6178cf : SUCCESS in 17m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/97d0175c137c43cfbb202788c2ce9820 : SUCCESS in 18m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e193efea2e6c41f6866c928dd8e7ad3d : SUCCESS in 20m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/07265f312867475681b53aed1a041dfe : SUCCESS in 1h 04m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55fa563eefeb4684b8c75a59e16aa4aa : SUCCESS in 39m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e8593602ed54e2783d35fe9712c0865 : SUCCESS in 42m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e377514b50bb43caa033175a27c6f902 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5af067f380249c091e1077b6285ad7c : SUCCESS in 8m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b752d782237424f80183b99291b0f9e : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8df587381a27cc805ff6ba1ad0bc33ea5944eb7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 07:43:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a38e3735a5a72794d783b17899baec53e69ec032","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 07:45:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f4cad22a4bc4d0195e548a78f6ef12d : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e7c11855f262437d816320113f776529 : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e639493776344daa971709f2c7e9c6a7 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e510728598a09cf4e85f9d653964fc3d626d726","revisions":{"1e510728598a09cf4e85f9d653964fc3d626d726":{"kind":"REWORK","_number":1,"created":"2021-02-08 15:37:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/81/774481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/774481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/774481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/774481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/774481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/774481/1"}}},"commit":{"parents":[{"commit":"acb742ac1fa2c233aedb1af97a0dd3ab1b5a201d","subject":"py3: Fix status comparison in ECGetResponseCollection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acb742ac1fa2c233aedb1af97a0dd3ab1b5a201d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-25 19:17:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-02-08 15:37:41.000000000","tz":0},"subject":"Prevent upgrading to pip 21+","message":"Prevent upgrading to pip 21+\n\nChange-Id: I952690558486a1d981985eae0fe9d7e07ff30e17\n(cherry picked from commit 0b870eb94dfe9769a07695b296eaa6758820a5f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e510728598a09cf4e85f9d653964fc3d626d726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e510728598a09cf4e85f9d653964fc3d626d726"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
