)]}'
{"id":"openstack%2Fswift~774510","triplet_id":"openstack%2Fswift~master~I4f900f81b446449ee699b4183ed061209709cf6c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I4f900f81b446449ee699b4183ed061209709cf6c","subject":"Add unit test for relinker conf file option","status":"ABANDONED","created":"2021-02-08 16:04:28.000000000","updated":"2021-02-09 15:30:03.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"471d5ce00343ad88b92d5e705fa589ece715998a","_number":774510,"virtual_id_number":774510,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-08 17:19:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 17:00:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-02-08 17:19:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f565ebc9c85fca59ffbbcd9ecadfbc75d36c205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-08 16:04:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29640e1d1e601b1fe5bc847e8be8b787a7954122","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-08 17:00:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cde771facfe06a930247bbc5946db809c392839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:19:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ed9c45f0ab4767b2e39aaac57632a7 : SUCCESS in 9m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8de829b6850d401da78f0d5e3367f5ad : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ccd76181ac444fae979f3a7b431793c8 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08650ca71ef64017b8f775d7572a0458 : SUCCESS in 14m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74d28a77e41d4f228ef55dc413a5e98c : SUCCESS in 13m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fe0d409322a1405e8ab50fdea492457d : SUCCESS in 13m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8e5acb76d374475bac5a7ec05fa5e76c : SUCCESS in 16m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a3bf9b678df4053bb0cb870ef8c31cf : SUCCESS in 54m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b10f4aedfcdb4894869c3b87d39d62e1 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec79826cafec4b18baf5802002640c6b : SUCCESS in 16m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/148271dd7b80467593aa2e0468d9655e : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2cf78d22ab034466b51ef642fe016fc5 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4b6ecd2ec7c54ee8af3fde9b8695aba8 : SUCCESS in 18m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e452635d8e1346a1b4fc3750f0aaa6f3 : SUCCESS in 42m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51ebb6eef12a42cb8b5a955af63bf821 : SUCCESS in 38m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70d0834985f14e378121ec7d25c40b74 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4198ffd91af64a9e9af02087812d664b : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec5e7400fc2f4e4ab831efa76314ee86 : FAILURE in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"471d5ce00343ad88b92d5e705fa589ece715998a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-09 15:30:03.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/773572","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17368c6aa39d8f022d918f5ac29e4ab80e29eab4","revisions":{"17368c6aa39d8f022d918f5ac29e4ab80e29eab4":{"kind":"REWORK","_number":1,"created":"2021-02-08 16:04:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/774510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/774510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/774510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/774510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/774510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/774510/1"}}},"commit":{"parents":[{"commit":"7294270291db331bacda72593b07e9a4f4cad92f","subject":"relinker: Allow conf files for configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7294270291db331bacda72593b07e9a4f4cad92f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-08 16:02:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-08 16:04:02.000000000","tz":0},"subject":"Add unit test for relinker conf file option","message":"Add unit test for relinker conf file option\n\nRelated-Change: I32f979f068592eaac39dcc6807b3114caeaaa814\nChange-Id: I4f900f81b446449ee699b4183ed061209709cf6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17368c6aa39d8f022d918f5ac29e4ab80e29eab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17368c6aa39d8f022d918f5ac29e4ab80e29eab4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
