)]}'
{"id":"openstack%2Fswift~774731","triplet_id":"openstack%2Fswift~master~Ia554ca7d0930d43e25973ed79dc0f16f3d3ba6ae","project":"openstack/swift","branch":"master","topic":"p-shard-compact","hashtags":[],"change_id":"Ia554ca7d0930d43e25973ed79dc0f16f3d3ba6ae","subject":"Shink the public shrinking interface","status":"ABANDONED","created":"2021-02-09 20:39:20.000000000","updated":"2021-02-10 14:37:47.000000000","total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"10438a706152d7485779756fa1b0af57274e8957","_number":774731,"virtual_id_number":774731,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-10 00:15:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2021-02-10 14:36:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 00:15:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 00:40:34.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-02-10 10:03:03.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"be03f75af67de9ed86d8272c8dd48fa3621f1b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-09 20:39:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"200998c09d4a431c2185416f669bf174d4710c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 00:15:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f5dabc012b45b6bc9b38368e29a62c : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e37a2c95292f4b8b9067a1e3500ff936 : SUCCESS in 58m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4d2f5d42e9142a19ad5fd0e28027963 : SUCCESS in 45m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5a52ad4a8134616bf3e1e22b08daff4 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/96f72ce24e0e4e258f53ffaae2329a97 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/161c51be5c59427184c9350a25f883b3 : SUCCESS in 19m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ee0c3b62a9a46c6b3c5673f2ab10ce1 : SUCCESS in 19m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/330a594448ae4918a991048c723384b4 : SUCCESS in 14m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/341bd47502d545cd9c2db9901204c14e : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f38dbd7db80541d4a1e43171d10630a5 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e36136dfac8a498684b8aa57ffd4ec4d : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0432c5385f97416d870f3d5533d123ae : SUCCESS in 19m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/986a9db9088f47399a2d4c61e057a299 : SUCCESS in 22m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b4776de240f8402fba13691b0e4a4bb2 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/40bd3f2aeb604bc09949da690b40d695 : SUCCESS in 22m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76e626426b9b4e74988a9e63b46fa4b3 : FAILURE in 24m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bce50c168291421e98aa451c988cf846 : SUCCESS in 50m 46s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/855d033cb07d484eab247789a61cae91 : SUCCESS in 50m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c5587180b5e4e58819453d71cebe8f4 : SUCCESS in 39m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd50517d5b704f928542ef8d4a893bf1 : SUCCESS in 36m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddc6c0e65b6544c5b3025de56cc02889 : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b413d7327b48df8e0b3cefbeae5c89 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/507943e03bde4bb8bc72b07dc21863ba : FAILURE in 10m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f37de43b02d41d3a77410309a2586bd : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d05f79ae9a6568ac641c0f2fc77104e42f4a0311","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-10 00:40:34.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2adeb67da8934dd4d7970d3547d0e3eed1f76267","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-10 10:03:03.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n+1 for fresh eyes, I had evolved both code paths to be...identical :/\n\nAt one point I had process_compactable_shard_sequences calling finalize_shrinking, but then I broke them apart to facilitate unit testing. Plus I have a fondness for using the same timestamp for a swathe of changes where appropriate - it sometimes help when debugging to associate states. \n\nIf we\u0027re going to write tests for the new public function (which I guess we should), then we could just as well put the call to finalize_shrinking back into process_compactable_shard_sequences and do the work to change the test for that.\n\nBTW, finalize_shrinking will be used independently once we have repair:\nhttps://review.opendev.org/c/openstack/swift/+/765624/15/swift/cli/manage_shard_ranges.py#638","accounts_in_message":[],"_revision_number":1},{"id":"9ab0cb2a1c24bc6243041e01a915141f68157bea","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-10 12:32:34.000000000","message":"Patch Set 1:\n\nAn alternative: https://review.opendev.org/c/openstack/swift/+/774848","accounts_in_message":[],"_revision_number":1},{"id":"ea6e3963ccd1fa96833c38a9553cf785c783f1c6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-10 14:36:54.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"10438a706152d7485779756fa1b0af57274e8957","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-10 14:37:47.000000000","message":"Abandoned\n\nthe alternative is better","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"210cb17393b124fbce2d46d95185ff1884b13655","revisions":{"210cb17393b124fbce2d46d95185ff1884b13655":{"kind":"REWORK","_number":1,"created":"2021-02-09 20:39:20.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/31/774731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/774731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/774731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/774731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/774731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/774731/1"}}},"commit":{"parents":[{"commit":"12bb4839f02cdf32170a85438973fd93da4c1140","subject":"swift-manage-shard-ranges: add \u0027compact\u0027 command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12bb4839f02cdf32170a85438973fd93da4c1140"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-02-09 20:36:01.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-02-09 20:36:01.000000000","tz":-360},"subject":"Shink the public shrinking interface","message":"Shink the public shrinking interface\n\nWe consume find_compactable_shard_sequences to produce actionable shard\nrange sequences, and return them to the caller to report and confirm.\nOnce the caller is ready to proceed there doesn\u0027t currently appear to be\nmore than one way to action on the returned sequences - so let\u0027s only\nhave the one obvious way to do it.\n\nChange-Id: Ia554ca7d0930d43e25973ed79dc0f16f3d3ba6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/210cb17393b124fbce2d46d95185ff1884b13655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/210cb17393b124fbce2d46d95185ff1884b13655"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
