)]}'
{"id":"openstack%2Fswift~774795","triplet_id":"openstack%2Fswift~master~I8167a1c62b03ece20f68a339c16645b2424e624c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8167a1c62b03ece20f68a339c16645b2424e624c","subject":"DNM - testing tempest change 774428","status":"ABANDONED","created":"2021-02-10 10:26:48.000000000","updated":"2021-02-10 12:43:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8359b531dfdb1b51113b54fbce15f63697fd1489","_number":774795,"virtual_id_number":774795,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-10 11:50:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 10:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bf674aa1b9ee3b88bba69d27a9f322d6350b420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-02-10 10:26:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ddfb5a1ab70521600838caa094256e0d04d7fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 10:27:20.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 499, column 31","accounts_in_message":[],"_revision_number":1},{"id":"4aeb9ea81404b88bab9ab1df7cfba82210b570fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-02-10 10:31:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7622362767e5178ce1b33df054cd7fbe142da74d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 11:50:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f0b541e745b47098df7fcc4d78cf949 : SUCCESS in 57m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b7d4ca3f6bc477cb620b39750010054 : SUCCESS in 46m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2cd0658e7c6c4ab3bec678ca98e424c6 : SUCCESS in 8m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/431633974792430086d90190138346d7 : SUCCESS in 14m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b736a2a12f1745cb91737a4359511f61 : SUCCESS in 15m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6410de0f63ac4242971cc8dca402ca1f : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":2},{"id":"8359b531dfdb1b51113b54fbce15f63697fd1489","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-02-10 12:43:51.000000000","message":"Abandoned\n\nnot needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5c9fbeb300417ce2b47014bf9ad9a2661ea5927","revisions":{"50199742f541c19d558e3e5ec1ce006a6eac2832":{"kind":"REWORK","_number":1,"created":"2021-02-10 10:26:48.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/95/774795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/774795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/774795/1"}}},"commit":{"parents":[{"commit":"eafeda8a5103f622b3166e83c4316ecb96d2fe13","subject":"Merge \"swift-manage-shard-ranges: add \u0027compact\u0027 command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eafeda8a5103f622b3166e83c4316ecb96d2fe13"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-02-10 10:21:44.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-02-10 10:21:44.000000000","tz":0},"subject":"DNM - testing tempest change 774428","message":"DNM - testing tempest change 774428\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/774428\nChange-Id: I8167a1c62b03ece20f68a339c16645b2424e624c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50199742f541c19d558e3e5ec1ce006a6eac2832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50199742f541c19d558e3e5ec1ce006a6eac2832"}]},"branch":"refs/heads/master"},"e5c9fbeb300417ce2b47014bf9ad9a2661ea5927":{"kind":"REWORK","_number":2,"created":"2021-02-10 10:31:49.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/95/774795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/774795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/774795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/774795/2"}}},"commit":{"parents":[{"commit":"eafeda8a5103f622b3166e83c4316ecb96d2fe13","subject":"Merge \"swift-manage-shard-ranges: add \u0027compact\u0027 command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eafeda8a5103f622b3166e83c4316ecb96d2fe13"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-02-10 10:21:44.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-02-10 10:27:24.000000000","tz":0},"subject":"DNM - testing tempest change 774428","message":"DNM - testing tempest change 774428\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/774428\nChange-Id: I8167a1c62b03ece20f68a339c16645b2424e624c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5c9fbeb300417ce2b47014bf9ad9a2661ea5927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5c9fbeb300417ce2b47014bf9ad9a2661ea5927"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
