)]}'
{"id":"openstack%2Fswift~775171","triplet_id":"openstack%2Fswift~master~I18dcb70159d5ae15368122dc6f9e7cd1221d94cd","project":"openstack/swift","branch":"master","topic":"relinker","hashtags":[],"change_id":"I18dcb70159d5ae15368122dc6f9e7cd1221d94cd","subject":"fixes and tests for relinker error_counter","status":"ABANDONED","created":"2021-02-11 15:00:53.000000000","updated":"2021-02-12 11:36:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ff32bf41301f8621a17c8a57358815696fdadb8","_number":775171,"virtual_id_number":775171,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-11 19:31:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 19:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f471caa182761b99a88651c972911d6e179ea45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-11 15:00:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c1ec9c97ff222dae37f0392753958ddf55d37fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 19:31:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4496e46a764311a1a3a11aa9feecd4 : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/c1a9add748e04b5c8208ebb06b43bb96 : SUCCESS in 1h 12m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a94cc8726d24e96a68b4cea1bf06eeb : SUCCESS in 48m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5fa421e4cabb4ad2a27927691525f9f1 : SUCCESS in 12m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bbfc4380a7d74283b3098d0618780934 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c187db27e3704f60bfcd1bf604780a75 : SUCCESS in 16m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70f01fbebad54deebd8d8cc1ba801f29 : SUCCESS in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63cf26d9f1d447e8aa85ad8e029453fc : SUCCESS in 15m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f6cdfda661b3493b9cef7d3ce1266ffe : SUCCESS in 13m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/591958a502f747f0accd583a92b9aaac : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5232a146ec3c45b1a066cc843c1559ea : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2bd78f541f3147fd818630c97a2dd403 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8284a34e8008471ca8e5d0218b351c85 : SUCCESS in 23m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b364519eb6ef4b11bab6502d10cc79d2 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/62d8f8777cb246eba855339d9c3fad27 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d8992742ed454e7f8dbe52419668b354 : SUCCESS in 19m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3f7eb40f5314f63873f6b7af2c6b970 : SUCCESS in 48m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7071606f35d542e7bf1cd7d8868362f1 : SUCCESS in 49m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f66a3ce391dd42ce911bb1dbc359e1a1 : SUCCESS in 52m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f618424707b44ff832a7f909ee9d83a : SUCCESS in 44m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f48dffcac0c74cd594f176b0c5a3613e : SUCCESS in 42m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83d7a9fc934c474b8020840b4512f35f : SUCCESS in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51729db1eb784bfc98c665c68af1b214 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71275a3ccd314426a8ed5974eb6ea631 : FAILURE in 8m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1016a801b44e4e53b20f81d323a027c1 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6ff32bf41301f8621a17c8a57358815696fdadb8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-12 11:36:23.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/773191/4","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b432796c47648a04898fb3c4c710ac7988f1c655","revisions":{"b432796c47648a04898fb3c4c710ac7988f1c655":{"kind":"REWORK","_number":1,"created":"2021-02-11 15:00:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/71/775171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/775171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/775171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/775171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/775171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/775171/1"}}},"commit":{"parents":[{"commit":"16a45186c1a43d862606fbf7fd1471b8762e327f","subject":"relinker: Warn on unmounted disks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16a45186c1a43d862606fbf7fd1471b8762e327f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-11 14:59:34.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-11 14:59:34.000000000","tz":0},"subject":"fixes and tests for relinker error_counter","message":"fixes and tests for relinker error_counter\n\nChange-Id: I18dcb70159d5ae15368122dc6f9e7cd1221d94cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b432796c47648a04898fb3c4c710ac7988f1c655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b432796c47648a04898fb3c4c710ac7988f1c655"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
