)]}'
{"id":"openstack%2Fswift~775224","triplet_id":"openstack%2Fswift~stable%2Fussuri~I7d869b137a6fbade59bb8ba16e4f3e9663e18822","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"I7d869b137a6fbade59bb8ba16e4f3e9663e18822","subject":"sharding: Don\u0027t inject shard ranges when user says quit","status":"MERGED","created":"2021-02-11 20:09:36.000000000","updated":"2021-02-15 21:42:43.000000000","submitted":"2021-02-15 21:40:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775224","cherry_pick_of_change":720786,"cherry_pick_of_patch_set":3,"meta_rev_id":"82df2a918637f2fa0da07b7337b5f70392b143a9","_number":775224,"virtual_id_number":775224,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-15 21:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2021-02-11 21:03:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-15 19:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-15 19:34:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 21:03:28.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2021-02-11 23:39:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f67568116580e264be905746a809c449ceff79a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 20:09:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bd1ec8c902e1e0c0156f9f276dc8d3f36974c500","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-11 21:03:28.000000000","message":"Patch Set 1: Code-Review+1\n\nSorry, I forgot that we\u0027re already on a release forward from Train and I cannot backport to Train and forget Ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"b22ef027fb59b4310ff786317b281f8953a11fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:39:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e52792486245ef9c0e96d8e23c51da : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/d8355a2ac4664267bd958239bc268831 : SUCCESS in 1h 09m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/242d51e62731407cac241fa3ba9bc172 : SUCCESS in 43m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dc9fd4bacf6b4191b88dce657aca0ef2 : SUCCESS in 10m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/910f48e08e9245fda3d24653200906cc : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e4b2d9a3e9dd446c9412d705374c4e4a : SUCCESS in 12m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc7e78902b324fa4b57e4937ba4a8745 : SUCCESS in 13m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/470b1fbe9beb475da91285a21041a2b1 : SUCCESS in 13m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50ac0731651147ea8a180eaaf590ade3 : SUCCESS in 13m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59a69b878c004b7da8778709c6084b60 : SUCCESS in 10m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa4ed2b9eaf0447bbc422d600039c8db : SUCCESS in 11m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b5e01ba422cf4c539f2c10c63e3f2c30 : SUCCESS in 9m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ebcc6a4d15f4e0d904bd435803e9aaa : SUCCESS in 10m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0e58c508a1f84a54a4e1e5ce857a2243 : FAILURE in 14m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8a937ebedcd74e1386b467086401cfa8 : SUCCESS in 11m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/462abf293cf249b8b1cc8e7bacf28694 : SUCCESS in 13m 26s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/992fe7c4ef504f9688c91e9ec2f06309 : SUCCESS in 10m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9ec4e85e68af4734b4e3b233fa4fadc8 : SUCCESS in 10m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4df6b1f1743343bba4de9cf0ccb6d8b3 : SUCCESS in 15m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aaf61b9f6631424285d316b807c6f549 : POST_FAILURE in 5m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0f13b41ab472419fb232b94b0a9f5748 : FAILURE in 43m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/792e289154e54359a21160456a85c41d : SUCCESS in 28m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/048a684af38c445ab729946adfe1a30c : SUCCESS in 27m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf34408942cd470882b9be9c48a54f98 : SUCCESS in 29m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb9560f6df63441bb53deca219b6014c : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa4f698ee6b4bb1b0ea6044367b0349 : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3ebd4e327084e0189bdeec16201b541 : FAILURE in 9m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e5dcb56f7b24e498a2bbb430fbd0ccd : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8e6f93ec017db32db9adeaee9eb840c280872165","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 00:14:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bae416019afab4d1afdbbc1f62c4fcef7fd5b123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 02:05:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f9027e7bf84ca7becf6f867dd87efa : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/281a587946844399a74b5f910fcdb0f9 : SUCCESS in 1h 10m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f53226a4b5f54549ab66e53a70e04002 : SUCCESS in 43m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/96daaf9dc63741bda4a98fb58448b997 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/42c392b48c1c4edca5e93ae475aad273 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e922637bf9249e7912e962f9f3d0e6f : SUCCESS in 13m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b255ea72fa2941679ee5445fdc4e210c : SUCCESS in 12m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f50a26c0af554d6da56abb985bbe0c7b : SUCCESS in 13m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/53f3d3a22fee4f3e92e3ead24ae7baaf : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1e67312db1914d73a2079460dba636b0 : SUCCESS in 11m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e969b663cc142b48b8d767998ce410b : SUCCESS in 12m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9173c30d18424328be9ce45ff5ae6e99 : SUCCESS in 11m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0a5930b2643d4b0da48a3dcc96791c30 : SUCCESS in 11m 05s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fd521e3c0eda4e4f983719bd3326473c : SUCCESS in 18m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/410db345e0f64820a270a963c6da7be1 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/342e2e7dea32444197f4a2f7707ee21d : SUCCESS in 13m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/21ab0a82be594367b07cca198a491c2d : SUCCESS in 12m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bce63698000f4f068522fe4a5af7926e : SUCCESS in 12m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/195f27a591cd4f34986fefdea6b4bab4 : SUCCESS in 16m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/38079988934d4e5886c38817eb98eca4 : POST_FAILURE in 6m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5f543b489d644f1d98105281550b6c45 : FAILURE in 50m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ac781bb73bc49e08dfc03839a2853ea : SUCCESS in 26m 37s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bab0f0571d0d48a79ff459d41004f33e : SUCCESS in 30m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5075a9f0dff4495d8ff572c51a52085c : SUCCESS in 27m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1431340462764efa8eac54a14a1725c2 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bee2d4a28d440d89c264c3d432bfbc : SUCCESS in 7m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f967a8444244b488bd7e6c2deb346e2 : FAILURE in 8m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e5c387b27c34d7e9f091258ecafa915 : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0db45f18b47f9dfa57a1591c3d41b38af026a5c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 03:03:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea518fefb81c3a6aa893ceb70ce9af1ba603bf97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 04:22:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd57be9aadd4ed98795e4802bd1375e : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/3438cb404d68483db7cf6a56980704c1 : SUCCESS in 1h 12m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75990415a7194fda9610dff4240b4949 : SUCCESS in 46m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53e09cc520c941e5a2835b1cc953f09c : RETRY_LIMIT in 1m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f2dd8fb46e81480a9dc028844805a803 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f489283e80041909ed700e261057674 : SUCCESS in 13m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/54ef4491cd8e4f78bb897dff375f73b1 : SUCCESS in 12m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c62b331be914253aac1ec1afe01d439 : SUCCESS in 12m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4fb505c2a2e64f2fb856e1c5d773bf51 : SUCCESS in 13m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62fbd55bd4c24ea7ad6ff4aa3f7aaf70 : SUCCESS in 10m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62895dec2bb1439f826ff11afddb779e : SUCCESS in 11m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dcdde73b3b174b99bd8b41bb9f5ad36c : SUCCESS in 10m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14a76edc99a24a079c37e849448aac8e : SUCCESS in 9m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/762a5d2df2d44ca8a3754597c4e35e3d : SUCCESS in 14m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5e14ef4849734bdf8370a5da955eba11 : SUCCESS in 11m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1e62eae4588d46e3aede8c5fa1fba19a : SUCCESS in 12m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/53993fb1aa0c4015aa5b0b4009d1c968 : SUCCESS in 11m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1de4b6c4bf584ae08bbe3c24a4466f62 : SUCCESS in 11m 12s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/216d7d1a8321475b970bc7f28221b128 : SUCCESS in 14m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2e915b1dbde45edb4d2dca212c5d2a2 : POST_FAILURE in 4m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a2348c28bd1f438baa0842c839c556b7 : SUCCESS in 40m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19df74ac92d74e0a9e0b68d02fed4778 : SUCCESS in 31m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5dd3f2a8809749008d1c8865d2928409 : RETRY_LIMIT in 1m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/610c532aeb634bdc91bfffdba9879cbe : SUCCESS in 29m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59e95eb66854465aad3ccc54386ce22 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c97144828200438188f99aebcfb236f0 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bc4debe4ed9543c28cc3a1a324406a97 : RETRY_LIMIT in 2m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/934ad18380534592b691e74a2ba453fe : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":1},{"id":"06839ae70b5db95797af28d7d26814c1b5c44b8f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 05:15:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94d95891cb92f093a9abb20ba209201dff5fa4ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 06:22:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe566af1f08741cd881a50dd57423153 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/cad81632388b4b258e575b4a40a86317 : SUCCESS in 1h 04m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba6d148277c64fd794efff547416b874 : SUCCESS in 47m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6bd71cca8b04b47973b3bc2cef1f110 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69bedb5c7c624e8aa74cc8df12c40e8d : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d03c7b7351f74d14a1649951963758fe : SUCCESS in 11m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17963707fd1c48a4a475b377df65005c : SUCCESS in 12m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5ffb1ab7e49c43d88f427215b5c6fbc2 : SUCCESS in 11m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/402f5f5c4b794e43ac545eab0e6baffe : SUCCESS in 12m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cadefa6b276841859bca5dffdcaff692 : SUCCESS in 9m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e39c8030e394f17aa7d605cac0ebc1f : SUCCESS in 10m 58s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b1d5bbe4ef68426686479291a5f419f1 : SUCCESS in 10m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d05f0d2a8d33442bb4a6458aa5b2a55d : SUCCESS in 13m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a409174bd460426f804b5534e74f1ac6 : SUCCESS in 15m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ced2f9d2402b4aaf9af7d7d3fc381d9f : SUCCESS in 10m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2f814f71ded84c57861eeb6a1692aa86 : SUCCESS in 13m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/37ada999860d452ca585c364e4095831 : SUCCESS in 10m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/59f5d54267e148b1bb705f7a6134ab5d : SUCCESS in 10m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b244d3dbca95478bb787bc22f60536ff : SUCCESS in 16m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10f14917dfcb43c1b63db9de915ef956 : POST_FAILURE in 4m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fd2c58c18c104c7399befc75b2dd9d6f : SUCCESS in 42m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6cc08e2d06404029b030abf5517cc0fe : SUCCESS in 28m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/64325545fdc5423f8ff1951a72b8f219 : SUCCESS in 27m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a58bc768afb4d5d80662bf472213755 : SUCCESS in 24m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f43f75038284217af40618eedd81f1e : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e0b28db52e4d638cd1609de3515ec5 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/929006251971477bbf8fd493fde140c6 : FAILURE in 9m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a5beb0bdffd4bfb94cb603274aa7462 : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"190d6ab10ec29dd8b14b9693006a1474c0ff94b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-15 19:34:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport; definitely seems backport-worthy (though I didn\u0027t bother writing up a bug for it).","accounts_in_message":[],"_revision_number":1},{"id":"ee8d3b8577fd9fe87a0bee250f21de7cb694e5ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 19:35:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bf901573ddd4b61ce6d08758e8342a9b00819d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 21:40:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d6c4b463f54630b3fec5a3289c8bd9 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/ee8ec302f7c441db9e1d6673ebe0e2de : SUCCESS in 53m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89b9495e39664107bf0b293579627090 : SUCCESS in 39m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d8ae5d25502f4fbfb2d5d2b03ed06ecc : SUCCESS in 8m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/95dd0dc26e2847af8b5b13418d5d5ffe : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6a67ff0c15c94ce8b0cc3062f3fbb037 : SUCCESS in 19m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b54ad7bf2ef484e81a95c1e679cd83b : SUCCESS in 11m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad0b13ab8ca74c998370a1df942ccfd6 : SUCCESS in 13m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/531289bdda9c4f9a8b578ee0534d759b : SUCCESS in 12m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc6b914ba9dd46aea544338ad183bba3 : SUCCESS in 9m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cad8f439ef8f486188e2a849e8cd9b06 : SUCCESS in 11m 30s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7e7029e6b0e44fc796cf356d2e08e70e : SUCCESS in 9m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9449211462f402cb9dc5a4003cd507b : SUCCESS in 15m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7bf3abeee0a243b9b92ecfb7d4185e12 : SUCCESS in 17m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/199cc901ac7245779e6fd2749b3ad969 : SUCCESS in 14m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/12e9f84497db49e88b804a855b7eec47 : SUCCESS in 13m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f50a3462d706475281a78d560e932b87 : SUCCESS in 11m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d3def53c04c742729c72d8f7c7163b0b : SUCCESS in 15m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95f7a85f3e5c4ec585e6c972b8b16628 : SUCCESS in 42m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97efd9b5d55f4610a2bbd33e81efe427 : SUCCESS in 26m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b3750bab86524467a8eeb5ba19ab9834 : SUCCESS in 26m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d387da6d99e84f078e8dea5a29a215e1 : SUCCESS in 31m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/812b8ba049bd416e8307d2a5b3b54869 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5019c483b88445496f9e2c3953da26f : SUCCESS in 5m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8ab6aeb9c664965af4204b11737c9a8 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":1},{"id":"daec2b0187052baab6a9805dbfca054ae64bf85e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 21:40:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"82df2a918637f2fa0da07b7337b5f70392b143a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 21:42:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33da4a01d5654ebba2cf42da8a328ccb : SUCCESS in 1m 24s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/023949bb9fe54b468c005f0c5451fc4e : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/38b2e224c0e94dbbb3453dcd017100ab : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a357fc66a72cfeaa41b9ef8261d5c4bf40a296bd","revisions":{"a357fc66a72cfeaa41b9ef8261d5c4bf40a296bd":{"kind":"REWORK","_number":1,"created":"2021-02-11 20:09:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/775224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/775224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/775224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/775224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/775224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/775224/1"}}},"commit":{"parents":[{"commit":"69a95a52a5a0dc383d30719be6d5c25273ae631d","subject":"Prevent upgrading to pip 21+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69a95a52a5a0dc383d30719be6d5c25273ae631d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-17 17:22:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-02-11 20:09:36.000000000","tz":0},"subject":"sharding: Don\u0027t inject shard ranges when user says quit","message":"sharding: Don\u0027t inject shard ranges when user says quit\n\nWhen an operator does a `find_and_replace` on a DB that already has\nshard ranges, they get a prompt like:\n\n   This will delete existing 58 shard ranges.\n   Do you want to show the existing ranges [s], delete the existing\n   ranges [yes] or quit without deleting [q]?\n\nPreviously, if they selected `q`, we would skip the delete but still do\nthe merge (!) and immediately warn about how there are now invalid shard\nranges. Now, quit without merging.\n\nChange-Id: I7d869b137a6fbade59bb8ba16e4f3e9663e18822\n(cherry picked from commit eae27412d233b6be5b1c2226a5d0ca380626f4ea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a357fc66a72cfeaa41b9ef8261d5c4bf40a296bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a357fc66a72cfeaa41b9ef8261d5c4bf40a296bd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
