)]}'
{"id":"openstack%2Fswift~775279","triplet_id":"openstack%2Fswift~stable%2Ftrain~I495fb1ec2394130c7274368662b58212ca375854","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"I495fb1ec2394130c7274368662b58212ca375854","subject":"Give unit tests a second chance to pass","status":"MERGED","created":"2021-02-11 21:39:44.000000000","updated":"2021-02-14 17:53:49.000000000","submitted":"2021-02-14 17:52:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775279","meta_rev_id":"6c18dd605d4cf703fa6fede5f40f3140e3e565ed","_number":775279,"virtual_id_number":775279,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-14 17:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-14 16:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-14 16:44:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 01:27:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d448ce4824163a8a4f84d7dda39d3b4363e4dffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 21:39:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d90fc43c0c3ce7fdf32e22fb5ad396509680d3bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:27:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71cdc859e7304b36aaa076aab86b4481 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb7ca3020a7d455e80f8cc9bf664f1a9 : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/15917359e4604212bf5fdb0bcf971737 : SUCCESS in 58m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1651adc6c5445a3aed5a6cacd29ac16 : SUCCESS in 42m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ed6df42f9ab4cf190e44d9d627e9d99 : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8748f41d7cc541c59ef1b9aeae39473f : SUCCESS in 14m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab92d88cda994f34b042058ad62047ea : SUCCESS in 13m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ab7d09d40424a08a0d62f51ccd10fc6 : SUCCESS in 11m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99bfdbe59dff49168bee37c3f7048084 : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d3585234bea4d9c969ea9311e77d3ec : SUCCESS in 9m 55s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/38cdbfe8520e4f4088d604b72552e9a3 : SUCCESS in 9m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7aa665f05f3646c0bba99f4dde23ae64 : SUCCESS in 11m 09s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f53af284e0ca45188de1a48fd35f1722 : SUCCESS in 12m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cca570e9d00c45cdb7432835ca9ecb52 : SUCCESS in 10m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/73377c82f77d47a79efd3d695d7ac8fe : SUCCESS in 11m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/800dbcb81b0c46de8f1b4f7900d9c347 : SUCCESS in 11m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a6e14f94ad4b4899918ec61d4a810df1 : SUCCESS in 10m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/21ff93a560dc4cbf85028b3aab2a78fe : SUCCESS in 14m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7793716f40c44ff99899be1c88e02a17 : POST_FAILURE in 5m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30b233fe06be4d3ea02c85e1800f5e21 : FAILURE in 27m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/88cb6c1853b94dfd8cd3a6a5b9e1657f : SUCCESS in 27m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b4b2e4bcf9fc426fa24b9063d260be22 : SUCCESS in 27m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b135e5971487436984301805096f3c7c : SUCCESS in 28m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/958852ba9000479bbf5b1f09882ee495 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e7222fe8c64348b44a89b4aa74e0e9 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/505f2c23dfc54803a00cd161865e06f9 : FAILURE in 8m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e6389c0c1414a07b6d7ec7e90c785a1 : SUCCESS in 1h 01m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5fd31e81eb1e40a8bf95293fd830afce : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3e8a2e652cec8dda7d071aeb8286287b0f50d436","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 16:30:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40f4496603a7b19adc8fee7d74b1c791609d0a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 20:33:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae8b5c22bcd4d92beb626af6faf4afd : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cdb11f9a994428581d474f8405b1f4c : SUCCESS in 5m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/bc8a7357371c478d9ab5eeb9c2355f29 : SUCCESS in 1h 15m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a05a1dfd8d8b41fda80203dce4e84daa : SUCCESS in 41m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd16b7b0d2c747c8baa2d23962e262e4 : SUCCESS in 8m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a44c8c20d334361ae913e187dbe89b3 : SUCCESS in 11m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0498e9ae4c104c178e831f3aeab8c49e : SUCCESS in 11m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/acd685b349434435a3f33fcc9a734729 : SUCCESS in 12m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f50ef16c12548ca9d61bceea15ca73a : SUCCESS in 9m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7eaddffa981e49c8b52349595e8d9473 : SUCCESS in 9m 53s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a61f82f70e174aeea2d5dbc9357ebfac : SUCCESS in 10m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/012192e6b4e447298d9d12924adb6a05 : SUCCESS in 34m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1f1b75eb53dd4e0f93ef7664cd1fe27b : FAILURE in 12m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/151a1f994cf743009ab280b3af94e4a0 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1904cdb4bd5643d69f82d3142b3a79b7 : SUCCESS in 13m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c359d795e61e49f384d0b922366f1670 : SUCCESS in 11m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b4d3032217484265bda8d6b0d65580c3 : SUCCESS in 19m 25s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/eefe47f983ff40699e1af99f5536bc3b : SUCCESS in 14m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a34762f6f5a24a51806ff9d876387c71 : POST_FAILURE in 6m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2c128813c0b9437c9942a8d267b65238 : SUCCESS in 32m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/504777de785b4b0aac901356b6f4f9f2 : SUCCESS in 32m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/86d2e7cd6f634c0e9f2db49f96b930af : SUCCESS in 32m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3a0c747482a04739a5907fdbbc8f16fc : SUCCESS in 41m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0b405ac93294cc284f5deb323fd47bf : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c846ce03eef41cc99e6c5d132eba5b8 : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eaa97328b7bd42459a56fca16cd31f20 : FAILURE in 10m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc0c90f59e93442b9319be336d5adff7 : SUCCESS in 1h 02m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04950b0c6eb6482f93d3ae8e8887fcb4 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f3372a3fbe6b54d124d0cd24a176ee417919a561","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-12 20:49:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00dd5e51d77b335f12f4d90ba4b4c1069fb5d9e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 22:34:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e52ba92d2ab497f888dbb9366b52066 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5479538dec204c7280b2537eec33723d : SUCCESS in 5m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/2f652a7c681943c99b6f72d8f0c3c718 : SUCCESS in 1h 22m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68d900afacb842788b50c593b529c156 : SUCCESS in 48m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68da31e5ed5146dbbf4bc485a89f8d94 : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ef539964caf43fe97ac9f8c7e524ad2 : SUCCESS in 11m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a05308a88231436eae6cf4db30a1d7cf : SUCCESS in 12m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/40077b1fdd9d4a338d268b2a7ebd08d6 : SUCCESS in 11m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d6938b1926943eba849adc61e081619 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b438aa2100a0441e97de3ce5adb34f99 : SUCCESS in 18m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c5899087b1424311b6bd82d0ab00d3ee : SUCCESS in 10m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7320ab121f7d45c19d756eed08840c29 : SUCCESS in 10m 52s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/af97b9bdb749486c9ae4c3a394137740 : SUCCESS in 11m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/51a98ee853e64057be5552555784d7c6 : SUCCESS in 11m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5660dcd9e62b4b458c4940fa8b5b4901 : SUCCESS in 11m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fa116bb67dd44ea3a45d0b778e658267 : SUCCESS in 9m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/de2455dcdef94c3db979f33e99f2123e : SUCCESS in 11m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ab9b3066478a4d1c8d7be47dcc404d9a : SUCCESS in 13m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85c38b2b50864ddb9691a502178c2b33 : POST_FAILURE in 4m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b15ad96edfb049ecb28416ba32846676 : SUCCESS in 29m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8321fc1058c548d3a68434c942085f2d : SUCCESS in 27m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/26d1b2e7ccd84218a836ea90f43969ed : SUCCESS in 30m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff10adbc44f841a9b9ab74a6431c2948 : SUCCESS in 23m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c192f247645e4a548dd799455d2757fb : SUCCESS in 21m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea307730225b49b0bd0047b48002ff29 : SUCCESS in 6m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45f947b5c0764db4a18b2a1902c3cb5e : FAILURE in 10m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8f4fe611ee14746913fb2565eadd677 : SUCCESS in 43m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9dc34078a7b3429ea23e4fcadea7df9c : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b41326558e11a1abaddab274e1259b0cee097a5a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-14 16:44:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3a9a29ee3173f4a342aadef4b42458505d10681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 16:45:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27629e9f22d65926d427380e850a9457a22cb255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 17:52:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0dfb312020748bd8febc8c109d3dfcd : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34405799bfd249e19dc759e73b8b1f84 : SUCCESS in 5m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/4db0e581b98944949444f1c55bd41621 : SUCCESS in 1h 01m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/56a2f635ba8243589bb3ca89ebcb5bf4 : SUCCESS in 42m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ed169d488d29464f88d6a29516cd2dcc : SUCCESS in 8m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c7a5b54e259405da004f493b9e87aa1 : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8864ed376c34a979888e39cd424cb1c : SUCCESS in 11m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/80de77cc20ee4fe584f879ccddbe3422 : SUCCESS in 11m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/390dcae14ba34635b3dd8536f2ac6725 : SUCCESS in 9m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/897dad8e9b0240ad9d4a0933396f0555 : SUCCESS in 9m 14s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4d30b49ada644fb8ab99dbe643f9e5ff : SUCCESS in 10m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3bc557a6b7c54f5f83b0ae2bef1c1259 : SUCCESS in 10m 14s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/650967c49fd3488db9a8d5e743b69b4e : SUCCESS in 12m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/817c8bc7558441c6934914b043e60e17 : SUCCESS in 9m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a63dad67218e402793a8e59e774baa6f : SUCCESS in 9m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/97148da5692d44939acb318a14ca2031 : SUCCESS in 10m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/20ac441e46e84248ab922ba294e985f1 : SUCCESS in 12m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52bc61cf4e604a33845bfdb8796c4590 : SUCCESS in 28m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f7ce4f4c2d1148eebcaf55f78aeec2ab : SUCCESS in 27m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c2c44992f154456b8cbe095aefa6e4ac : SUCCESS in 30m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d42226c5f634c2986f693bb0445721c : SUCCESS in 27m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7723cb6407549b4be660f2ac5a36538 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b70a2810350424b83ec80ee223f4d5f : SUCCESS in 5m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31e2ab6c02424104822d8438a2970cbf : SUCCESS in 1h 00m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9bad6f0e3e1447abbb034304b75459b : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6ec6d96b984e06924cbfa05fd8e24db86addea0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 17:52:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6c18dd605d4cf703fa6fede5f40f3140e3e565ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 17:53:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ac342b95cb49bb852f45c4d138f203 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4bb7870bbb249beaa47a162d41f86b1 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2b5c4121a4bc43d2a543de888942a8ce : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1e15d15ae505987227358364362db7f0c0f5b06","revisions":{"c1e15d15ae505987227358364362db7f0c0f5b06":{"kind":"REWORK","_number":1,"created":"2021-02-11 21:39:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/775279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/775279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/775279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/775279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/775279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/775279/1"}}},"commit":{"parents":[{"commit":"fc742d6027b9036b66b6eb64b5015365fd342969","subject":"Prevent upgrading to pip 21+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc742d6027b9036b66b6eb64b5015365fd342969"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-04 18:49:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-11 21:39:00.000000000","tz":-480},"subject":"Give unit tests a second chance to pass","message":"Give unit tests a second chance to pass\n\nChange-Id: I495fb1ec2394130c7274368662b58212ca375854\n(cherry picked from commit 976cc8f482723a698160815affabb540e4d766eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1e15d15ae505987227358364362db7f0c0f5b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1e15d15ae505987227358364362db7f0c0f5b06"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
