)]}'
{"id":"openstack%2Fswift~775282","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ie05521f6cd146234dc5615c96ad19681b43e9110","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"Ie05521f6cd146234dc5615c96ad19681b43e9110","subject":"Give probe tests a second chance to pass","status":"MERGED","created":"2021-02-11 21:46:44.000000000","updated":"2021-02-22 06:22:21.000000000","submitted":"2021-02-22 06:18:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775282","meta_rev_id":"e6db81cfb64e9e7eec3452c610f266fa5c6b6b6b","_number":775282,"virtual_id_number":775282,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 06:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-14 16:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-14 16:42:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 01:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cf555309f4ca8a1a4f14de59d2a7c8b3904348d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-11 21:46:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01b24ca792ad65b8fb4883ec2f8eed743f295364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:36:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42866bf0f62b4cf7956428871ab0842e : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/efaaf3984a984205b83045cb088727dc : SUCCESS in 1h 09m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e82a5a358ac7479b89f21fbdff87aa0b : SUCCESS in 45m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/558af501c85d4ef9bc7c495d28fe01a1 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/394a6b1792854f4cb7a6fe9f979c4313 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c09eb8ab8102424693e1e1205c8f35c8 : SUCCESS in 14m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6fb25676b6c4970a81fff9ed0eb687a : SUCCESS in 13m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/26f3f47f86f14652aa9fc7c049543fac : SUCCESS in 14m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eabbb899f3bb4938b9f98c7fb5bfc9d0 : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19ea3b3d241f4156bc72222a53e50365 : SUCCESS in 9m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e46c6bb5f5e4f47ad428456929e16aa : SUCCESS in 13m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4bf7df7225c24eb8b4901f3e77c8b4f7 : SUCCESS in 9m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e76d4556a6c449ab9ed0f2cdccaa78d9 : SUCCESS in 10m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c02255bf96b9409cbe4d981f05b56682 : SUCCESS in 15m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8f063203657b48e7877427e40db96d56 : SUCCESS in 10m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6d19a31a5eac4a9ea7e6606532d0ef16 : SUCCESS in 12m 28s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/18313e8d75e3400092c50f6e6e2ab5bf : SUCCESS in 11m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/928eccf7ceff4b20a1901d5125df28de : SUCCESS in 11m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4131fc59d4f14a2fa8c6c65cf998f391 : SUCCESS in 40m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba9f00d2191848f1996981b533e77f21 : POST_FAILURE in 5m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fb6006f5efc748388ab7c4b6b2ecfb57 : SUCCESS in 48m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c906a66d53241aa9618dbc2cfc05aad : SUCCESS in 30m 59s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/93d0257b9ae74ad69688350d5aa29ede : SUCCESS in 28m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae6260c8cb414930b6d0f3cb8a3e9ea6 : SUCCESS in 38m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/186f2752db4b4fb5abb392e1ab73eb47 : SUCCESS in 25m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99fc8189e54946318ad1d0d4e65fd6ef : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aafea96face94967a0ddd3b37b61ff88 : FAILURE in 8m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abcbe5d2d6c34861b9621684d775bce3 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2388575fff875629ae5afe4f77a693a9d6348488","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-14 16:42:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e27102dcc6377a9107f153109b3ccc52f6c55ad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 17:58:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b8ec73a899fdc68913d48c0daac4ccdeeae2200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 18:51:45.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"cd11fa45768be916a669fe17d96501cecfb4cfd0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-15 06:57:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7446765a0d7ae7ece0fb0cd911c084889ee554e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 07:56:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f465fe5b983749948d97725fce3c0edf : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ee74ffd426734df59e61a64afe341e21 : SUCCESS in 56m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e4a6e8f8dff4a0ca647bcbeb9186eb8 : SUCCESS in 45m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f0dbe329a804ae3a49c830c9a7c3f63 : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e39542834184adb8e5ddf8c38bf657a : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f33450b5cd414d8ab6c536aa7949b0d6 : SUCCESS in 15m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1078e0523cf4e19b2448983293f2f3f : SUCCESS in 12m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/df97e3859ada4677a10fd1f6c384738d : SUCCESS in 12m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8031b869493449d8a2deb952f4da593a : SUCCESS in 13m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0105e92965804f77836511de8fcee5f9 : SUCCESS in 9m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f29a6fd4aaac4487aaabf5c28a727ddd : SUCCESS in 17m 08s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/baf9294369c3409a81bc6a59baea1e9e : SUCCESS in 10m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d1925847a37f419890372362ae55694b : SUCCESS in 11m 25s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cca9491ff6a14096a74bbeda22f0b766 : SUCCESS in 14m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e9bf96cf0ea4cb89bc76d171a515021 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8f8ca6bc87004767ac890dd7bcc11aa6 : SUCCESS in 17m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/52bc330be7654fb39772351be17440b1 : SUCCESS in 9m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/715d5783ff2443d3ab468fc195e36798 : SUCCESS in 10m 40s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/20c4608819924f198dc85e37f309364c : SUCCESS in 22m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87ba5feef3d24d61b2fe52d584a1903b : SUCCESS in 26m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cfb1b1a64f3c4a20b569071d8e947c7b : SUCCESS in 53m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05ccecd817c24e8ab75014320355f824 : SUCCESS in 23m 15s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6c8b5eeb9f5841839cf950a4f3dcced2 : SUCCESS in 28m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9db30e20f0f24ad28b5ed6a7d8391e5f : SUCCESS in 27m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbe01b1907b54883972b8cfa50f24665 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3293b560ab94a54a92ea6e13a3dc277 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f4ce4c846724586b6464b9d57040fc3 : FAILURE in 8m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a27a3cdc868e4d0eb1391632c0b67ce0 : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":1},{"id":"77a6ccf2bab4460ff98f71b8862b8e4767deb7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 08:11:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3d4604942e673bd858aa76ec127b71e82df541e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 10:04:20.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"cb966bf1c6baadfc32a01fa101c39be4fcf58fea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-20 07:24:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b33a6fb8ad3f3f897f60fa0e322e03ec59eabde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 08:45:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce47eaa577a84da1b3ad1b3f1097bcbb : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/2215dc2da3ad4f109ec9e77cae119467 : SUCCESS in 1h 13m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fe906771d36543e7a5daf2bef0f32ccc : SUCCESS in 39m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d00719ca1c4547b99d3223a1d42f02b0 : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/59e77d374f22476ab7e930f2abd74050 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7001510f5aa3473dba869cc35e9ab0a4 : SUCCESS in 15m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f36b4e331e44f358fc1474e89e22c4d : SUCCESS in 14m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/819ef1e4f0b14436b64a021e594645f5 : SUCCESS in 12m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4d3eba95340a49e8bbb8638b53da17f1 : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b27237e223949dda5ae3b2f9cb26efb : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/12c6f23e93ac43a9a30305ed6b5288e4 : SUCCESS in 24m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7b78b070b424490c87fff7a83434614c : SUCCESS in 21m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ea5480d3072f41a78a6f5cd237e8f6e0 : SUCCESS in 10m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/171350ff566143a68d70710aed2bbce0 : SUCCESS in 26m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/51f9154787b9493ea0531bc4b7acace7 : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/250baf2d44b8431e8e9261bf2390545b : SUCCESS in 14m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c196f3f39f7f4309b9e50856fc158acb : SUCCESS in 12m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fb0908e149454c4eab36149be3dcfbb5 : SUCCESS in 22m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7fb9228d298747188d8cda3652b6310c : SUCCESS in 18m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7438b79e56024f8a9852e8df826ef662 : SUCCESS in 27m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd94cb1deab54a99924c75b6024d98e4 : SUCCESS in 42m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/528a97fbc212425f9c9b66e7bc0ecbf7 : SUCCESS in 31m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/538fc8a9107c4bc3ab53ef89dece775a : SUCCESS in 24m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e0e7bef71824a9b9289fa22671f6933 : SUCCESS in 31m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfeb94090213466ea6ebcae4d8a1d861 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a538eb5361ce4e529533e6ffc728b57f : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8aa8ca46ef6447fa8aacaa23547dd91 : FAILURE in 10m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60111efc015b43f4b878290d827434fa : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":1},{"id":"baf7895eb3f03bb534d4f1f35241aa73bd7be337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 04:28:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8baf3ea72239a2645790b438100e1b9cfe03cacc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:18:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2fc0eefec6747579d22656e74d30b4c : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/94f3cec799c74049b8901aec3cb1d93b : SUCCESS in 1h 11m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0eacb6291d934e6f952bffcc0e162b2d : SUCCESS in 37m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/04c7268a87ce4e9cb316407c98266441 : SUCCESS in 9m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/625c251bb95640a49c093c39a9618c0c : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd575dab1b774b919be6f650ccda1a6e : SUCCESS in 13m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/067719b9152142d7a4f505da4a9a28bc : SUCCESS in 17m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b2616416f42b46079c0c8e86b192f537 : SUCCESS in 12m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9f947b58b9e843f68cae43160498e497 : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3d9de41f4f8492c8bb5db0d4ec82201 : SUCCESS in 10m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45c2589b34bd47b0b72671943dfb8dcd : SUCCESS in 16m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6f648c55bc6d4cac9b206c88e1a3f5cf : SUCCESS in 10m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3bb3adebcc8045bc942ef71f5f143b64 : SUCCESS in 13m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/738cf976ffb3434ba360539c7d8740d5 : SUCCESS in 14m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/55827c5adafb49b8ba08a8f44f01bd0d : SUCCESS in 15m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ed70a55c5fde41bd8a1bf5202de8d687 : SUCCESS in 10m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/83f829d0606e406ea642900f45a74a5f : SUCCESS in 11m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/382217404d6040bbb3dc84c43f6a237e : SUCCESS in 23m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c99c4d4feb6c4f269a3ae7558949c25a : SUCCESS in 41m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f88110d7d57b443394327d3f262b1dc0 : SUCCESS in 31m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f5ef76468ddf4c2e97fa670289ec1cef : SUCCESS in 26m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d01ec47b704b48f896d58a53c8efcab5 : SUCCESS in 34m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14344957eeff415a9e2feab4af8d5f01 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e325eb243a4eb0a32c2a1000d57ca1 : SUCCESS in 7m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bcf2f3ceb2744a68d63dcbd2b918765 : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d1c0012b40bb0c834044d987f2e30fcab9df72a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:18:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e6db81cfb64e9e7eec3452c610f266fa5c6b6b6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09c49f17fbd4c0c9be5c6c8714a716b : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f09211ca84de4f8d87fae36bd9e56497 : SUCCESS in 1m 10s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c250c4cdae974d35b22fe5e299db3ab3 : SUCCESS in 1m 08s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9491d73a639f99cf7260241a0860a41f68f5bfd9","revisions":{"9491d73a639f99cf7260241a0860a41f68f5bfd9":{"kind":"REWORK","_number":1,"created":"2021-02-11 21:46:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/775282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/775282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/775282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/775282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/775282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/775282/1"}}},"commit":{"parents":[{"commit":"69c93dcb16de217f196d2423a49012bbb17dae52","subject":"Bump up probe test timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69c93dcb16de217f196d2423a49012bbb17dae52"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-12-08 23:21:36.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-11 21:45:40.000000000","tz":-480},"subject":"Give probe tests a second chance to pass","message":"Give probe tests a second chance to pass\n\n...and bump up their timeout, since that seems more likely to happen if\nwe have to retry.\n\nChange-Id: Ie05521f6cd146234dc5615c96ad19681b43e9110\n(cherry picked from commit d4c0a7d3b3106f8b491e78ea21fca36c99ad04d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9491d73a639f99cf7260241a0860a41f68f5bfd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9491d73a639f99cf7260241a0860a41f68f5bfd9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
