)]}'
{"id":"openstack%2Fswift~775470","triplet_id":"openstack%2Fswift~master~Ibe1d18d94f9d5af7654a41e2086ea1a7b42e0750","project":"openstack/swift","branch":"master","topic":"relinker","hashtags":[],"change_id":"Ibe1d18d94f9d5af7654a41e2086ea1a7b42e0750","subject":"relinker: Pass whole conf dicts around","status":"MERGED","created":"2021-02-13 05:29:13.000000000","updated":"2021-02-19 05:22:55.000000000","submitted":"2021-02-19 05:21:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775470-relinker","meta_rev_id":"a45cc88b2331f1719514dd3496dbc99310641242","_number":775470,"virtual_id_number":775470,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 05:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-02-18 19:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-02-18 19:55:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-13 06:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 19:48:02.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"187a71487776733506776009929e294f07e00995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-13 05:29:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3848af54f826597804b30b55710e44354eb18eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 06:39:12.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7071bfcb8f34465eb20d38033e9f0a81 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/39c61b25c3d943558d2f239c5d826af9 : SUCCESS in 1h 05m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3cac170434c44dc38a60718db8d81f4e : SUCCESS in 49m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7aa09ddba0e4edcaaa6a4543a52ca87 : SUCCESS in 9m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5cf097786e646b9809bc7debc79c487 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28df80b01da74549977ede09cc86dcc6 : SUCCESS in 14m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ec7c4ab57c3472f98afae2925636372 : SUCCESS in 14m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be525ef365e84238bcc54031baf2cbb6 : SUCCESS in 14m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/79440c3c42e441298948e684de3a48b6 : SUCCESS in 14m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b42dfe5ca25f49d1823d971d4a866e69 : SUCCESS in 13m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/58c747d9197e44e581d076a040e5625a : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5eabf83988de41aa827a0758a7426d4b : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d1d3b33a2ad44d74a91cbe94949d93ab : SUCCESS in 16m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bf3bdedf9d81457691df96efa67e299f : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b1a3c18d70c14f53ae7ff7c2ebef3d02 : SUCCESS in 20m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d08f2c5160648efa11f035257bb3f48 : SUCCESS in 17m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c282d1cef226430e980d9e280c72799b : FAILURE in 21m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7a9454a5d2948fa82af6c43465c0018 : SUCCESS in 50m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7041918b3bad41d8947f381a8af9b8c8 : FAILURE in 48m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/324ccc82cfcc462d88e2a09def86788e : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2fc94b1a7994a4d9fef54b9576dc50f : SUCCESS in 43m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa2edd64d1114782ad7b4127f2743e5b : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e639489fc345f79701425443343d4c : FAILURE in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8e6dcdbe57d14840a9db6b0a2f26daa6 : FAILURE in 7m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/529fc2f30f5d49dea532fcfee765b2b3 : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0651c67b2b7d8324d1944b91efb78705de442719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-18 19:48:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3da039e755c2c32986a2d3a3f92095dd4c9c4501","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-18 19:55:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nmakes sense. extra test to cover DiskFileRouter getting the conf here https://review.opendev.org/c/openstack/swift/+/776521/1","accounts_in_message":[],"_revision_number":2},{"id":"15d689ea951cc1955e24b7153d79ed9de6196a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 01:03:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bfcbab3449b4443bd01045f9d060206 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/2f70fd37b4d043e3a792a0d658bdb35b : SUCCESS in 54m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/043abae12b8d4b23a742f1f59512e8d4 : SUCCESS in 40m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2aa541c6a8bc487a83e39515c324ed73 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f5ef53efd4be44b99fdc2ef4d1be3d7c : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/297c60db700b416cbe6993f2d5114f46 : SUCCESS in 13m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8b4845e5c4464b099e4453ebe0d92e01 : SUCCESS in 14m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2639b63c927749e9b1641573565fd195 : SUCCESS in 13m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/980d3bd3bf6b4151b28e2a7c6c6db3c7 : SUCCESS in 13m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54370b533d3b4e90b9ed089955b3221d : SUCCESS in 15m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/493478f28f284e43a249d6449059dee0 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a45520b02384c4f845309b3361788b8 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61fb4ae5d28c4fda8ca186a4b7e0c7f2 : SUCCESS in 19m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b7e6e1166f9b43349a14bbdcb6255862 : SUCCESS in 25m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e1585e2957094d888f39e94da040d7ee : SUCCESS in 17m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2e8f850aefeb4e2db3d2596776cfcd0f : SUCCESS in 20m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a967bab1502d4844b15acec7f5eb5fca : SUCCESS in 19m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e7e59443d8340109f84ca53482f97c7 : SUCCESS in 45m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d93288f6b8074be9a91a223176e5a896 : SUCCESS in 46m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3da21108b67749e3bb900feb509f55b2 : SUCCESS in 37m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9294907b8cd496ebeea630d78623a8f : SUCCESS in 44m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1243c0d641074d11900a66905d35f29b : SUCCESS in 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5eab8e38b324e62bfd35f34521de2ce : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f883a9874fae45c4ac25c44bee764a91 : FAILURE in 12m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c13c305cf73343fd807308c54013b8c0 : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"cb17b0814f117f1308cadaf10522c978f78016a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:56:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81d43074c8fc7768f9aaff514656050655545488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 05:21:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e10bddac31e49799aae0d6ad4bf4bdd : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/0ed7ec01fa7b4864ba6667716b2589e0 : SUCCESS in 1h 14m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7fdf08847a914dc993eb973c11dd601a : SUCCESS in 45m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8992991022584f888b3cfcab68c43e92 : SUCCESS in 11m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/22e710299a104079b660561bc331e9b5 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a8091834df74427a891fe8a83bf582f : SUCCESS in 15m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b3a9644569b41198017e6b71fc789f5 : SUCCESS in 13m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a4050111821b46e7b39628ad061babfd : SUCCESS in 14m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a11b7da03534422981e567c92666eee6 : SUCCESS in 14m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/27bc30d6dc164458a149a46f541e440b : SUCCESS in 14m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1e88d5d2f4384d2b87341b04a11d7e06 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/39ecd7c8b3d14057b3f0589cb0cc0fa4 : SUCCESS in 14m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7978b21307834051b8be23b4cad69f33 : SUCCESS in 22m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1cdf29f3e22043ef99f3e2ee20e4044f : SUCCESS in 22m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/30acecaae34340209e06b97bf48c0371 : SUCCESS in 39m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c1e9770dd015498f98d7f1c4c750c9c0 : SUCCESS in 24m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4351db8a11ce4ba8a5899e1fd35e4bb1 : SUCCESS in 54m 04s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/261281e41fd1486aae5b8ce7477ffeae : SUCCESS in 45m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f71730a9be8d4f79a81c82fcee0cf818 : SUCCESS in 40m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/536ec519a08b4e8b8f039d546613df82 : SUCCESS in 48m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dd2d7c510d348b4ae9c776fb13ad65f : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f13de3e1c14c2a9d87c95ba37eeefe : SUCCESS in 6m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bbb9fbc1609496699350610e4aa8548 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"00c3ab46b18158f8efd1747575b925a4a2bd550c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 05:21:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a45cc88b2331f1719514dd3496dbc99310641242","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 05:22:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf3213ee47f34a69934aa6d0a6daf937 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9955196bb0bd46cba3d30bdfabcee4b4 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ecb627a5504140ae8ed7f19f3b6e2d3c : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b30d9e818f87a6eabbd6a3c199ce34c3e23c070","revisions":{"af30f6d3b7b24c13f2537101fb8bd5598f72e9f2":{"kind":"REWORK","_number":1,"created":"2021-02-13 05:29:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/775470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/775470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/775470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/775470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/775470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/775470/1"}}},"commit":{"parents":[{"commit":"178c5ae456bc7056c6660008cf6d9e100343ef9c","subject":"relinker: Rehash as we complete partitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/178c5ae456bc7056c6660008cf6d9e100343ef9c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-13 05:25:43.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-13 05:27:40.000000000","tz":-480},"subject":"relinker: Pass whole conf dicts around","message":"relinker: Pass whole conf dicts around\n\nThis lets the diskfile managers have access to things like reclaim age,\nwhich it important now that we\u0027re doing rehashing. While we\u0027re at it,\npass better log_level values around.\n\nChange-Id: Ibe1d18d94f9d5af7654a41e2086ea1a7b42e0750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af30f6d3b7b24c13f2537101fb8bd5598f72e9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af30f6d3b7b24c13f2537101fb8bd5598f72e9f2"}]},"branch":"refs/heads/master"},"6b30d9e818f87a6eabbd6a3c199ce34c3e23c070":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-18 19:48:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/70/775470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/775470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/775470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/775470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/775470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/775470/2"}}},"commit":{"parents":[{"commit":"d24884450f6e852a6d9626e45892ca9208783cb9","subject":"relinker: Rehash as we complete partitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24884450f6e852a6d9626e45892ca9208783cb9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-02-13 05:25:43.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-18 19:47:38.000000000","tz":0},"subject":"relinker: Pass whole conf dicts around","message":"relinker: Pass whole conf dicts around\n\nThis lets the diskfile managers have access to things like reclaim age,\nwhich it important now that we\u0027re doing rehashing. While we\u0027re at it,\npass better log_level values around.\n\nChange-Id: Ibe1d18d94f9d5af7654a41e2086ea1a7b42e0750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b30d9e818f87a6eabbd6a3c199ce34c3e23c070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b30d9e818f87a6eabbd6a3c199ce34c3e23c070"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
