)]}'
{"id":"openstack%2Fswift~777279","triplet_id":"openstack%2Fswift~stable%2Fussuri~I133c2524460ae9df779077d160ed6fc2858c08cc","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"I133c2524460ae9df779077d160ed6fc2858c08cc","subject":"py27: Suppress UnicodeWarnings in ShardRange setters","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-25 21:19:39.000000000","submitted":"2021-02-25 21:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777279-ussuri-sharding","meta_rev_id":"41773b9d5d3755b07759dedaf2c913903cdc0e7d","_number":777279,"virtual_id_number":777279,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 21:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-25 19:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-25 19:50:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-24 04:33:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1cd9d9a77c73d1d1fc3a53f254653b5fecaf46e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6618fc0bd77ed4edba803551bbce253d229b9d6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:33:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be50e561c5d42f1a9c15fe12963ad38 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/143c1ff06dd245ec9edf1ce0d9433aa3 : SUCCESS in 1h 16m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b909934a50f94a839206c2ce09aacf12 : SUCCESS in 46m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0047366d05294eb7b4e1a9389d9aad33 : SUCCESS in 9m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e25ab692d36c4c8ba075a4e495231572 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c272ad6ee68462a97025758091334a1 : SUCCESS in 12m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/42c339f08a5c443183cf020e6fa3a0cc : SUCCESS in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d3bfed6e7294f41b8501d34ba41546b : SUCCESS in 14m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb8d0e279e404eb6a45d18985ffc48a6 : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c1aabc08e41415a99495c6a767df522 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/002c1d2ab04b4a8687feeb7d5a1d00e4 : SUCCESS in 13m 10s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cda1aaca507b4301b250fc5f7879a240 : SUCCESS in 10m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/483a0d7b0ead43e3a5410b32e5438baf : SUCCESS in 11m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3744c0aeb3c04bd5940c370ac178c27e : SUCCESS in 19m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b10bcc07ce1c45d3988c899c4f731b6c : SUCCESS in 11m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/92488375f7334ce0962aa92212a8717f : SUCCESS in 13m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/29be3275a3344663afaf2e6e3404c132 : SUCCESS in 10m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/951e87fe154943e4ac190f7351d82fb9 : SUCCESS in 11m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/be50e8eb0255430682f3fbfeecb00bec : SUCCESS in 16m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40d0cf3fbdfe48cf928f7c22b909679d : SUCCESS in 26m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dbf973da82a045978756892626c164e8 : SUCCESS in 46m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4147b25ed9354a019b3bcc8ed1ad0198 : SUCCESS in 32m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d9766b4af047491798c46ba8dc1f9d0f : SUCCESS in 32m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b28748c4a874bad99bf3be9f7f0a18a : SUCCESS in 30m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70e16685e7524899afc273c35b5c68b5 : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b062402b54347238d215f1bce9e3c61 : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50eb951331694ce586f3d76992b94b3d : FAILURE in 10m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e12d7c1f8c3149aa8de059b69e91d726 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"53d2a8fedc79dd55b7247f4407a1ce1a6d30a018","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-25 19:50:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, tests cover the fix (on py2), originally shipped in victoria.","accounts_in_message":[],"_revision_number":1},{"id":"4c8769bf68a07309cf908773fd29c024be67eda5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 19:54:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17e18d34262bfc7cf3a316d4e74079f8a0857fde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:17:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeb4ee6f40d644fe974168041aaa7603 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/f935633ff26345e88ee595a805e680b2 : SUCCESS in 1h 16m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26b20333d0ef4f3c9c1bd79e9c655220 : SUCCESS in 47m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/59d3e7bb5cf448c58adcb5e11b89106c : SUCCESS in 10m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9020d9628b9744b79ffb1dd8589da69a : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0727374cc8f94af7829c55c6f11b4091 : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/250df2c4e68c449ca8145ac8b7acecb7 : SUCCESS in 14m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/475590eeaf9044039380488a09db363a : SUCCESS in 13m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7666cf4ee99c40598b1ce0606a0e22db : SUCCESS in 14m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a24d55880174435cb42029d83fdbba01 : SUCCESS in 11m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d83fe9a153f24230bb4fd358cd505361 : SUCCESS in 27m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bff0f1318c8242e69110ed7b9b51f773 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a2c495dc830403aa6bdec60f3663ad2 : SUCCESS in 17m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cbd371c36c3e4490a363209a051f4be3 : SUCCESS in 32m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0a7ae3b9dbf241f7a86d9bf10b2958ad : SUCCESS in 14m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2636d80ad4014193a7cbdcaf310ad2c0 : SUCCESS in 15m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9503d674a8be4e95be84e838db3149dd : SUCCESS in 14m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/49781897f288473f8459702380ccb23b : SUCCESS in 22m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/903e6e3df2964ddca0cb476312fccbab : SUCCESS in 45m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/469ebac7f56d49978c1e52fa8a69881d : SUCCESS in 29m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/aa06aef27c1b45fe83f736a17a940902 : SUCCESS in 29m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/05f717c1fcf64436934a686237d2d8df : SUCCESS in 29m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dc64e38be504a70818ccab279f582da : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ce44d45e0e444fb3a6260bc558b0d4 : SUCCESS in 6m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae5b1439fd354fc7a2e34093b2a73eca : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5fd8b4455790827653cdac5b1c4aec517ca9ba48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:17:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41773b9d5d3755b07759dedaf2c913903cdc0e7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:19:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8a252dc08c491fab6c82cb55e9ce1d : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3a8322a8cc644cbb8592768bfc5364ac : SUCCESS in 1m 09s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6e10014cb9ae4b25b725f22a7d3b4230 : SUCCESS in 1m 10s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1b997889de06f8bcbeb38d0192fa4391465bc47","revisions":{"d1b997889de06f8bcbeb38d0192fa4391465bc47":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/79/777279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/777279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/777279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/777279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/777279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/777279/1"}}},"commit":{"parents":[{"commit":"1b3b776964651acae6c17f581855b0fc7094acee","subject":"sharding: Add probe test that exercises swift-manage-shard-ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b3b776964651acae6c17f581855b0fc7094acee"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-05 04:35:58.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-23 23:52:25.000000000","tz":-360},"subject":"py27: Suppress UnicodeWarnings in ShardRange setters","message":"py27: Suppress UnicodeWarnings in ShardRange setters\n\nPreviously, we\u0027d see warnings like\n\n   UnicodeWarning: Unicode equal comparison failed to convert both\n   arguments to Unicode - interpreting them as being unequal\n\nwhen setting lower/upper bounds with non-ascii byte strings.\n\nCherry-Picked-From: I328f297a5403d7e59db95bc726428a3f92df88e1\nChange-Id: I133c2524460ae9df779077d160ed6fc2858c08cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1b997889de06f8bcbeb38d0192fa4391465bc47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1b997889de06f8bcbeb38d0192fa4391465bc47"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
