)]}'
{"id":"openstack%2Fswift~777281","triplet_id":"openstack%2Fswift~stable%2Fussuri~I17a495b76a6e4072cfc47501e0f892fcf6c56173","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"I17a495b76a6e4072cfc47501e0f892fcf6c56173","subject":"Don\u0027t auto-create shard containers","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-25 21:43:14.000000000","submitted":"2021-02-25 21:41:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777281-ussuri-sharding","meta_rev_id":"7082a230113e070ac65a9e007618246b4de11420","_number":777281,"virtual_id_number":777281,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 21:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-25 19:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-25 19:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-24 04:40:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d16cacac827e3e4dbef1e1d978c1d3343ce8071f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"811c2bb22c72efeaeb6955a298c47fe82fb4a9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:40:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16d524ce4b3497395d9b72a7aed825d : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/41a79f3cc6f84136aa692647ec9e4f9e : SUCCESS in 1h 17m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba780e5b6d5244a79402a4ddfa2ab341 : SUCCESS in 47m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d732281c1b046cd8651cb9091428d4c : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6d73e1ceb0ff4f4392b6894c58941d65 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab1704af9c6c421d80076733d2abab63 : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e054a28e0764b45998e443b55f5521d : SUCCESS in 14m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/38b1e22ffa28452b97009df1c439fc9a : SUCCESS in 21m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a338fd99496a41738603f97d34acd8c4 : SUCCESS in 15m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ddad958f76643caa8b6d2423470feab : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/640a839413c54c9cae3214d0c5903637 : SUCCESS in 26m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1bfc26e3e9d7448793f32c9d57ee4033 : SUCCESS in 10m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e616725d85f46c192449643b44d4189 : SUCCESS in 21m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/69ec8766c2294fe9812e6b487bf5d42c : SUCCESS in 16m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d52e192a5ef34e24b74cfc3961589268 : SUCCESS in 10m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/12c4844e31b249b2a7e049d869403abd : SUCCESS in 27m 05s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d1ac2ac3d8194e589287ff8d9ece0afc : SUCCESS in 11m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f34c3e6e5d7d4d81ba4d3c3d6f5590bc : SUCCESS in 11m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5a207bedc43a4e2faed93de3be3a65d8 : SUCCESS in 30m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04226a1c1f544530925636afa08f1988 : SUCCESS in 36m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf240c33212847e1b53bb4a2caf4dbcd : SUCCESS in 46m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62d63eb4672d40879356a7e47f94897e : SUCCESS in 38m 38s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b275e3932bbb4b58a0a18164f911a36e : SUCCESS in 28m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ce9ef4b7cf14520acec39f09d1f8b18 : SUCCESS in 29m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df484068d87e4218a6caae1afd4d2c8c : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7b1f0b847b43849baba7c586d0dd3f : SUCCESS in 7m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f01d15182a543dcb6796a2cc7e5a7f1 : FAILURE in 12m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/087412f614a547beb5f7a959c737863f : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"06374ca6763f4fd5de1a1faac3b4432bff365f07","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-25 19:55:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, tests cover the fix, originally shipped in victoria, closes bug.","accounts_in_message":[],"_revision_number":1},{"id":"a29c43209b118bac58a5ae6be80efd48da3ab142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 19:59:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d46975d5854f3e200c04c3d47d2a395be940c38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:41:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3959640063944f3a9e3a05b3a1bcb22a : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/dd44cc37f5b5491cb8388e330028c503 : SUCCESS in 1h 39m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68011dc00eed421a9e5cfb5013df630b : SUCCESS in 1h 01m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3b5aecb573b245208d2a23da0772e28a : SUCCESS in 10m 04s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a6f1a8db624945c1a9e0989b0e233ece : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2eda5e45f004875a69d0d9ac19229e7 : SUCCESS in 14m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d4ee38c9afa457c80f570b3d615ee5c : SUCCESS in 16m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7fa4cb37b1aa4e4d974ac1399bdb7ade : SUCCESS in 14m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3d6130231e34478a8c96c08bf3ad7b5e : SUCCESS in 14m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cf7e1b549f4c48679b3da72939399033 : SUCCESS in 12m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c2397993e6741d082a6af79114eb359 : SUCCESS in 15m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/451b3298e46241fda38b3fceb0617d55 : SUCCESS in 11m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/de054b63d3a445dfbf5a651bd2f8d49f : SUCCESS in 11m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/79584df6e6454064835e9f98926cf569 : SUCCESS in 16m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e3b59ff6ba9e4c7299d726fa991dae9c : SUCCESS in 14m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ce0ffeb625f6403a9594b2dc46706786 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aea5974a14cc4cdd874824567585f048 : SUCCESS in 12m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/60868401c7af4cd69411907d97b80187 : SUCCESS in 16m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1e410bf78970443ab68e18c93b81fe3d : SUCCESS in 41m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef4e50b5a2b54d65b14e9fe714a9e7cd : SUCCESS in 33m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ab9f4ed6b1154ac59290242f1fa2bacc : SUCCESS in 33m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5acfc25bc24a4399aa703302e64cf227 : SUCCESS in 33m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8021259923984ca9acec931ba3a39d70 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a706d22573441b7900f862d84a7dc23 : SUCCESS in 6m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd3fb46744b1494f867ac46db90d5dec : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2747d975767fa4758d658e86dfb2382d3214f70c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:41:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7082a230113e070ac65a9e007618246b4de11420","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:43:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26147b169c2d43f28ade1868d1a8a3b7 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/38a88e797ba1418894e321590e58e91c : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/00d872e7ad6544f6a4dd0233e44e7229 : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8588296dcfbbb52ae007e2a299e042e09c1fef4e","revisions":{"8588296dcfbbb52ae007e2a299e042e09c1fef4e":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/81/777281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/777281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/777281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/777281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/777281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/777281/1"}}},"commit":{"parents":[{"commit":"0da122383b4fb86d11e21406df46311999c87461","subject":"Latch shard-stat reporting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0da122383b4fb86d11e21406df46311999c87461"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-29 05:20:21.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-24 01:28:41.000000000","tz":-360},"subject":"Don\u0027t auto-create shard containers","message":"Don\u0027t auto-create shard containers\n\n...unless the client requests it specifically using a new flag:\n\n   X-Backend-Auto-Create: true\n\nPreviously, you could get real jittery listings during a rebalance:\n\n * Partition with a shard DB get reassigned, so one primary has no DB.\n * Proxy makes a listing, gets a 404, tries another node. Likely, one of\n   the other shard replicas responds. Things are fine.\n * Update comes in. Since we use the auto_create_account_prefix\n   namespace for shards, container DB gets created and we write the row.\n * Proxy makes another listing. There\u0027s a one-in-three chance that we\n   claim there\u0027s only one object in that whole range.\n\nNote that unsharded databases would respond to the update with a 404 and\nwait for one of the other primaries (or the old primary that\u0027s now a\nhand-off) to rsync a whole DB over, keeping us in the happy state.\n\nNow, if the account is in the shards namespace, 404 the object update if\nwe have no DB. Wait for replication like in the unsharded case.\n\nContinue to be willing to create the DB when the sharder is seeding all\nthe CREATED databases before it starts cleaving, though.\n\nCloses-Bug: #1881210\nCherry-Picked-From: I15052f3f17999e6f432951ba7c0731dcdc9475bb\nChange-Id: I17a495b76a6e4072cfc47501e0f892fcf6c56173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8588296dcfbbb52ae007e2a299e042e09c1fef4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8588296dcfbbb52ae007e2a299e042e09c1fef4e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
