)]}'
{"id":"openstack%2Fswift~777282","triplet_id":"openstack%2Fswift~stable%2Fussuri~I56b4169002122ba3e53e78abd1ad994e0ae5ed3d","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"I56b4169002122ba3e53e78abd1ad994e0ae5ed3d","subject":"swift-container-info: Show shard ranges summary","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-26 01:53:53.000000000","submitted":"2021-02-26 01:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777282-ussuri-sharding","meta_rev_id":"c683573e6f7416705fdf9e8127ec9a3bd0fc47e0","_number":777282,"virtual_id_number":777282,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 01:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-25 19:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-25 19:58:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-24 04:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c735dfbf5dd430418277403c133e88733061d50a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d4708c119f0942731218dd76abb341181d8722d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:37:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fcf85df594401c935a4cbb04f506c9 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/67d7e02dfb684494897fe1e3d25c1435 : SUCCESS in 1h 14m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ec2cf130fd44bdc8dbd8b7699a67be1 : SUCCESS in 1h 07m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1261a49d72fe4cac93828ad0f7be0a44 : SUCCESS in 10m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/376cb9048ae947cba68e9ac1ce87fd1d : SUCCESS in 8m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7b854893ad740bf8a1b3ed8fb9b4b75 : SUCCESS in 19m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4be06704eda949e892b3a1f3fcdec82e : SUCCESS in 13m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/75383791cea44727b109c4eda5d7b142 : SUCCESS in 14m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/257b623038d64f1b82ee89e329e28222 : SUCCESS in 20m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/041e9012a21a4d6a80637010c383a94b : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/112779378abc4ea48c6ca3873dae2724 : SUCCESS in 12m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f66f75f54add43d5bde2d7891bd529b1 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec34d9e36ae547f2b8638e448b86295e : TIMED_OUT in 40m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/733ca14848b745398d5847e360f3684a : SUCCESS in 15m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/be376659164d43fc839c8ba4f4b8625a : SUCCESS in 11m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5a6c89f6c8db4636a5c65a8ab55938ec : FAILURE in 15m 38s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5c154e1709b7404b92db0b810ea9b0dc : SUCCESS in 10m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c59ec07959484760818cb911fcd61ce4 : SUCCESS in 13m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9a42536cc1834b9fa1102cfed14e6e54 : SUCCESS in 18m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/28f6acc8686d4288bf7a686ae01b14a6 : SUCCESS in 23m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/57d534c8abcc4a59914c1db72b98aef9 : SUCCESS in 42m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3209927a4f7b4133884054f93d5e6f5e : SUCCESS in 24m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7b45a7520d024b67bcb2671e32eb8e1b : SUCCESS in 47m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da7cabb85bd14403ba9e2db1e065bc61 : SUCCESS in 28m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6c4ec5409c84a8fbea54d7b205194c4 : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f63c8eb31e46198e19c8af1a47079f : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c285c4077871445593ce72089df22758 : FAILURE in 12m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4db2ddb9163748cc886d44585ec57c6e : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2c03d05b49729430cdb630eadd5f503ef20e875b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 14:36:15.000000000","message":"Patch Set 1:\n\nrecheck\n\nThis looks random.","accounts_in_message":[],"_revision_number":1},{"id":"602f0b79da7b0d8e79ed84007cd77bb63dab1e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:45:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc89f7205f9d416189d63e1d0807cd85 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/0499d9df4cbb41519538152efc33d321 : SUCCESS in 1h 15m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6221a63a44e54070b4e4e2e6d6f114f3 : SUCCESS in 43m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e06d522784d4fcaa5212fed6b09aa71 : SUCCESS in 9m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6adba7ed3490435893a51a54f99901b3 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0cab3d33e3c8485fa730ed0deca29fc3 : SUCCESS in 14m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c4eae84f70a8406f93fb71e831cbd483 : SUCCESS in 14m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/27a4e0c746f84eff9c9564c05eb995d3 : SUCCESS in 13m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8eeb40ac51784d948252f2fd1ad8e7f0 : SUCCESS in 12m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/67acac0d35594e339a233bf8442b0bce : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd2a9fd2aca4451c8bc446dfb953257e : SUCCESS in 34m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/74a74d45d43e4afcad8464c31b25a46c : SUCCESS in 11m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/69cbd8dabd45450cb7fffc8840d2b553 : SUCCESS in 12m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1ef4eb9842bb4a5cb43431bd98a54cf3 : SUCCESS in 16m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/77b87e63b5d742468547f4fd287f5c20 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aa2180b973b4497c894cc4dc9c7f0989 : SUCCESS in 13m 34s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b0b156272267491783e1bb0342141b73 : SUCCESS in 10m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c7178d92d17d4e7a9da334f218fe6323 : SUCCESS in 12m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/dfce5d91d9c24b5faa9c057b4bb14c5c : SUCCESS in 25m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0004dc37dcdc4fc692f2e6c79adfbbb2 : SUCCESS in 25m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fa09e7be76d14f0480b8bf700ac78813 : SUCCESS in 49m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fbc5884c7c1a492aa5ce7fc02f5f82a6 : SUCCESS in 27m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fbe9cbe71f3e473fa83e3ff730ea7c55 : SUCCESS in 25m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4435d2f2a28049ffa0d50f32394a52e9 : SUCCESS in 33m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f293a41865d143e4a781b6368905e32d : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c2f892216e47f09e6cf664dd0980c9 : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb5d71459622443e8de1e095c0c77039 : FAILURE in 12m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de15194c5ebd4e1cbd9a7d85fe7d39b4 : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":1},{"id":"50165c5b7af65877b3a7f20dc766ef85f07b83e8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-25 19:58:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, tests cover the change, originally shipped in victoria.","accounts_in_message":[],"_revision_number":1},{"id":"193035e49dad5e4de2dc5b12bba78cd0d36ae745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 20:00:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbe5108710e9a3d88b3de3c04833de3eada3e0ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:41:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8882001b297846f48031ac7202c2eddd : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/5525e5a0d5754a05b4f8d38e78680e87 : SUCCESS in 50m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1cba9fe91354c039cecb201b56a3362 : SUCCESS in 38m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3acb1ef99c4a4551866e9838483f61d9 : SUCCESS in 9m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f36dc08a74994d148b8a4b4f70da1208 : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af24b61706284dda9df666a53f9721a2 : SUCCESS in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6268358f109c4c38a3d3f26d7fc2b1ba : SUCCESS in 17m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed21954720474f5587834fb120d84296 : SUCCESS in 24m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ac98da74e5dc41e09122411924e8f5bb : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/88124286711846d1b5aa9f11df7b772c : SUCCESS in 11m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/de35d44952684274a538800ac2f6c3a7 : FAILURE in 13m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/32379416511b458ea6362c71367559d6 : SUCCESS in 10m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e466a0473d574b189726afd276dcbc81 : SUCCESS in 10m 47s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ac41583acd7e433aaa3b2057626240a4 : SUCCESS in 15m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/98d691dcf7ce4d68b1b51d310f613de4 : SUCCESS in 10m 04s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0938a86758ba46b597db8256251a5828 : SUCCESS in 11m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/697ff97dc0794a88893457d672b20bee : SUCCESS in 12m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/77aaaeed9d944c7db507930088c77286 : SUCCESS in 15m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/72ec8e8002fc4b83ae01fb0b30c7c8d9 : SUCCESS in 41m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61837a3d32224281932f55d2ba63d006 : SUCCESS in 30m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5b870f03f3cc4f7c9b4aa14bf558b0b9 : SUCCESS in 30m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9ed78cb2aaa4e63b37c3ee906ee082b : SUCCESS in 32m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dfeb363d43040aebfc75d44ef1b654f : SUCCESS in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f89101ff55b6481b9aad9d323896bdb6 : SUCCESS in 6m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dd3522b69de431dbba90449eb9718ef : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":1},{"id":"984e3d704137fb3b42d69e25a96707c7228552a1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-25 21:43:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec1f4782a5162253da12cef9417273897d8e73c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:44:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68d6b244168409780eb99d79a23fc80 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/25ece458942b4b43b649266cfcc0865e : SUCCESS in 1h 14m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba4b1c27a16b47738030423d69049724 : SUCCESS in 36m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6607bcddc9f468c9a2186ba6885ea7f : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/47b977a9241746049d3ff92549e61543 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c10aa61f7c05492185cbeca54a3dce3c : SUCCESS in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d19f3a50cc249618b65257a1d74245c : SUCCESS in 14m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4870b960c704aa79d643bd73f56b1b5 : SUCCESS in 15m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d80d8f9763c5460aa7b3e0cae84e56f7 : SUCCESS in 14m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/15312b034a9247a8a9db21fb9d8f5108 : SUCCESS in 12m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ddc079eda9d409c9004f410bfab9202 : SUCCESS in 13m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1154d4fbbafc4393a2f9b2ae83754404 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb7a3647dbc645b9aa9287299ab0dcdd : SUCCESS in 11m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/179ad9120f434693916dcc7a6bbd8919 : SUCCESS in 14m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bc98ff495a4c4a9fac031a43efacaa77 : SUCCESS in 11m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5a8a9d9df2fd430aa15c9fd93da4d62c : SUCCESS in 16m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8414b21abfe043d49169da24bbe3e7f3 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/35d3f50b402c495f861ade42de4ef8fe : SUCCESS in 17m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a8dc31e3911a4f90a46ca47ff6c957af : SUCCESS in 18m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da26f041bda74363b59767300c391781 : FAILURE in 22m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e07a40a649ba4f99b7b362bdab1daad5 : SUCCESS in 47m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/941de557bea54435b5575e85e36d9ea9 : SUCCESS in 29m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1bc85448bc7a48afb6beec10a4978aec : SUCCESS in 33m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58c565b4660e49689835c0017efcafe4 : SUCCESS in 30m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aee869ff1da4e4faad8ed9e1bd51902 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd13f6294ad451d8317dc0cf676ac7f : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5aabc4583b3c4df5b1722bfb1f5704a0 : FAILURE in 14m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1cb01450534f738e387d1819f2830b : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d2dfcb8f372ae0348feb9132fb4b25aa00b9bbd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:45:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d3e1495c90a1739a9b8d21209f496c1b6bc2632","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 01:52:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9962bd8fe8a54b28a11df49d9b9732a6 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/bfc034c84cbb410db41eb42f966dc582 : SUCCESS in 1h 05m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/509f97d9361b4300b4dcfb5aa0cbf36c : SUCCESS in 38m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c084c094c1d0426ebf4842174db9c4be : SUCCESS in 10m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f7d2b01266124acd8107d716768bffb3 : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4cc0875a71e401bb9a53e06a59a289e : SUCCESS in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/225524f8d3ea4edfb6808e7874af898a : SUCCESS in 18m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/42cf4a1c6b1b443fa1f0580d9c7cbf97 : SUCCESS in 15m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9442eec61bdc4e45b013db0add4cfb8c : SUCCESS in 18m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a34b130b853849a8bd631f4ec4480cfd : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7503fce73c8749999e758546fddabd55 : SUCCESS in 20m 08s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a830c9e8a7db415c809d272b84a9b6dd : SUCCESS in 12m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/94d95a2394504b24afe0fcddc3023c69 : SUCCESS in 17m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/69d1aeee125a45b28ac9a45e184b2cf7 : SUCCESS in 16m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bb55e2f736504f3297fc22ef3b1c80d1 : SUCCESS in 16m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0484f2c24a734524a5e8d9b1ad35ad08 : SUCCESS in 12m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/370d93ba4ad440db9111df9510e05239 : SUCCESS in 17m 12s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0b6153ef007e43a796c891f8556f548d : SUCCESS in 18m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7ef8376f68c478fb8135c68d3ac3bb4 : SUCCESS in 45m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46c9ed9aba7d4ae1bb2cc00ce62d0668 : SUCCESS in 30m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4b0fb7153e2b446d9f2bbbbbc455c44f : SUCCESS in 32m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e15041d85f2942c9bb95003af5878b4a : SUCCESS in 31m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee54580c97944402ad1423f3e5778a4c : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6489427720a24290bb38a6245ddc5f86 : SUCCESS in 7m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd8556b63dcb48518db33e5ef4380b34 : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"82412fca05526fa5775e9fe1b6dca69850cb46d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 01:52:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c683573e6f7416705fdf9e8127ec9a3bd0fc47e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 01:53:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd1968e1bb549898df8ee4db41d0463 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3c9b1a06dd6843fab68fbd7ad38cfca8 : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b16cc66b22a04327b396d275f60c3833 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4061c04364adf8933841968b053ff442f95059e","revisions":{"f4061c04364adf8933841968b053ff442f95059e":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/82/777282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/777282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/777282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/777282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/777282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/777282/1"}}},"commit":{"parents":[{"commit":"8588296dcfbbb52ae007e2a299e042e09c1fef4e","subject":"Don\u0027t auto-create shard containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8588296dcfbbb52ae007e2a299e042e09c1fef4e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-19 18:09:11.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-24 01:42:44.000000000","tz":-360},"subject":"swift-container-info: Show shard ranges summary","message":"swift-container-info: Show shard ranges summary\n\nThe current behavior is really painful when you\u0027ve got hundreds of shard\nranges in a DB.  The new summary with the states is default.  Users can\nadd a -v/--verbose flag to see the old full detail view.\n\nCherry-Picked-From: I0a7d65f64540f99514c52a70f9157ef060a8a892\nChange-Id: I56b4169002122ba3e53e78abd1ad994e0ae5ed3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4061c04364adf8933841968b053ff442f95059e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4061c04364adf8933841968b053ff442f95059e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
