)]}'
{"id":"openstack%2Fswift~777283","triplet_id":"openstack%2Fswift~stable%2Fussuri~I7b7be3b62e033ab668316e3c698ba667ceef31f8","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"I7b7be3b62e033ab668316e3c698ba667ceef31f8","subject":"Identify deleted shards as shards","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-26 23:39:06.000000000","submitted":"2021-02-26 23:35:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777283-ussuri-sharding","meta_rev_id":"0a2fb812752fe2777aeef4ad10654de1bd3d15ee","_number":777283,"virtual_id_number":777283,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 23:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 20:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-26 20:15:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-24 04:35:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bea20e073f7413a1d9084c35ce9e29dbe2988c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95e705b53a58b638e8771b89eebfb9c9d7321a98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:35:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9309ed527ac408e8cb903a31a8b53b5 : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/5f13d244e2a04899b1b3be637fe7f84a : SUCCESS in 1h 15m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08b7cbec36bb4dde9311597044b227ec : SUCCESS in 35m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bb7d5113aa3437b90897b76ff768fad : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/72e8c0b067a34aa1bf418101be5899d7 : SUCCESS in 8m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/788269235a17408080d3961e4beb1ccc : SUCCESS in 13m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eb04a70743e24f3eba136636ef430156 : SUCCESS in 18m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fa7f3c9c89414d49b1ae2923fe907b61 : SUCCESS in 13m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b1328c20aecb4e07a5b643112bd1d03c : SUCCESS in 14m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7385a35ff1d54d1a9961b8779d42a165 : SUCCESS in 10m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff0fd7bfaccf48fe8791e916a34eddc9 : SUCCESS in 18m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ad6e2beec86c441eab5f315c1c062cb5 : SUCCESS in 10m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30fb7601a18f4fc5b13db5fb911fae55 : SUCCESS in 25m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/b960f27fb26a4b64bffc9a78449a18a3 : TIMED_OUT in 41m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/863c9e93f9fe4b87a22bc421b636299a : SUCCESS in 10m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0a07334b22a74a16a4248a990e589d5d : SUCCESS in 12m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6de62a28340640558e4e24da028b3c21 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/60d9ca64f67a4753a680a586e5501516 : SUCCESS in 22m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7c623b9ef5e74294843bb3e80ff111c1 : FAILURE in 38m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb7cf346aa95437b863e0e18fac5952c : SUCCESS in 23m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6c59352f46584f7cb9e3edd40662fcf4 : SUCCESS in 43m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28ff771fba5244f98fa35e30989cb600 : SUCCESS in 29m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/871137c967b84b80b2075776edc8261d : SUCCESS in 28m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a15094f813834783942f73997a5927d5 : SUCCESS in 33m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05e7ec64127f4c1bb9eb7556c542728f : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25109ee127484468b81793efa04bc119 : SUCCESS in 6m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b33e666a00a412086717a3d8f3bdd30 : FAILURE in 12m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c9078ba48a04e5b9f308e0c2434e7ca : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dd2c626f359428ff01f253423c122e7027115e37","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 20:15:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, originally shipped in victoria.","accounts_in_message":[],"_revision_number":1},{"id":"228ce98e5aedadc63b5eeb5dcc8c464e68643c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:36:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e7f7a455c144898062997ad0360f82 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/9ee92e3d21df424e9017847f471cde6b : SUCCESS in 1h 18m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba4756af79a94423b75a8b4a8bdacef8 : SUCCESS in 49m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f281819ddf774dbbb30f4daba9002732 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/97c31e94f2d44ab7ae487eb8fef664f4 : SUCCESS in 8m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee77d7cf560e488daf6a472025483761 : SUCCESS in 21m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ce3256b6d91442f6a9cb87771acc68ad : SUCCESS in 21m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ee230e02b87e4a22ac9da03bd70ab6a0 : SUCCESS in 16m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/051d3d0050ee43c297e2401f31bb557a : SUCCESS in 16m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/88e1d4a27c524407bbfd56f9d940edce : SUCCESS in 11m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a571ad15f5ea4f358f11f63fe94f83bd : SUCCESS in 13m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/dc5bedac6f024667ae09cc63a2e09cd0 : SUCCESS in 28m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fb955129f66d41e68c297fc57834d762 : SUCCESS in 14m 13s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/72846aad64e7438a8b721a618d4c94af : SUCCESS in 17m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8a0a48af09754967a52cef892bddcdb3 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/20fc908b0594486e8a027dc872adafe2 : SUCCESS in 37m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/263b333dc5ec4acdb28e7a644f323f33 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6245d1a993294a59895478b895e93b3a : SUCCESS in 19m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8c7a7f3d5d6a44c3b08b730f77ff22a1 : SUCCESS in 28m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/972ce03db5384a38999384608fedf7c2 : SUCCESS in 29m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/82e999bd7449467e8779703be553284a : SUCCESS in 1h 17m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37d6c5cdc20b4ddba9a2ab1936d12674 : SUCCESS in 24m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/985e7d782ca14ed9a523588501ad194c : SUCCESS in 42m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa9245c2b6874717b90c22e8afa407ee : SUCCESS in 45m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e17d1c8f905a4c33b716933b61736faf : SUCCESS in 30m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3edf4ca2b4c7481fa02eef8cf40cfdc7 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e7ccb7a713f4d529bfdaf15d520e11f : FAILURE in 13m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b83ac4ac02834c3b8d831d086f08c768 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2da2dc3d0f0133ab3ba56e0c728bde711967900c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:36:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3de8a1bd9a9844d0ba5ea8b1e0b50d6642a96ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:35:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb59e1b15b841feb0b1ca4f85fbb0f2 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/58bf2bf2fe6043b4a67f37d3f1c42797 : SUCCESS in 49m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edce1247f4384bc49aa856f1475ec25f : SUCCESS in 35m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4724f8e6b4864578b43bb14422cb22e9 : SUCCESS in 13m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dc87e1485b654de2862f71b62c0c7c92 : SUCCESS in 8m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e1012c598244e1c8f4d120fb2fd7f17 : SUCCESS in 17m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6d6042a341ee4cddb8d0f90062e9134b : SUCCESS in 13m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1146a5a6b374402ab1318fe1adb5e57f : SUCCESS in 25m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/23b99dbae2504067975df5d186471dbe : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd0ecacb49ba4b0c96425cf173cc769f : SUCCESS in 10m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f575ad5eae2c4c6eb1dd12717dd1d7d4 : SUCCESS in 10m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ceaf2a36288142cc94d348ecd6b8a800 : SUCCESS in 10m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d313e9d30d7f4d9584b78a7e886897fb : SUCCESS in 10m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f1052f7bb8a64a97907acd90e3b0ca01 : SUCCESS in 26m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/01558ab23acc49edb548342901c53926 : SUCCESS in 22m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/db683790029b484ebca3f5436e7fcf53 : SUCCESS in 22m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/044f47a99cb8435b862887f7526ea696 : SUCCESS in 24m 15s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/51191bfa20b145488ad3b6a48ddaae45 : SUCCESS in 14m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/82896d4c48944e3fbe1e95a08e69d12d : SUCCESS in 42m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9ca8df38f164106a1301629c18ba3fc : SUCCESS in 45m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/df4c526f63614a4d806e808425b577ef : SUCCESS in 30m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc641dcdbb404fb4a2324e652838a0c6 : SUCCESS in 45m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f11adfac543b4768b6004e26b547c6a2 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a1841294fd441696fef0439841e6b1 : SUCCESS in 5m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58ffff880d41487fa2f1e2f71405efb7 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8f60f74fe434bed8b7f63a27f6e699a5e039a4a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:35:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a2fb812752fe2777aeef4ad10654de1bd3d15ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:39:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c964818a1e21413683c0f40cfe07b973 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9bcbc5dfe56d4722a822effc8a6b4645 : SUCCESS in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/148691b2bce84bea86e1939eac5d817e : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eef68d69c586f9231a20b24d973893b154353290","revisions":{"eef68d69c586f9231a20b24d973893b154353290":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/83/777283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/777283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/777283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/777283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/777283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/777283/1"}}},"commit":{"parents":[{"commit":"f4061c04364adf8933841968b053ff442f95059e","subject":"swift-container-info: Show shard ranges summary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4061c04364adf8933841968b053ff442f95059e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-25 20:42:01.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-24 01:57:36.000000000","tz":-360},"subject":"Identify deleted shards as shards","message":"Identify deleted shards as shards\n\nWhen a DB gets deleted, we clear out its metadata. This included sysmeta\nsuch as that used to tell shards the name of their root DB.\n\nPreviously, this would cause deleted shards to pop back to life as roots\nthat claimed to have objects still sitting in whatever container they\nsharnk into.\n\nNow, use the metadata if it\u0027s available, but when it\u0027s not, go by the\nstate of the DB\u0027s \"own shard range\" -- deleted shards should be marked\ndeleted, while roots never are.\n\nThis allows us to actually clean up the database files; you can test\nthis by doing something like\n\n * Run `nosetests test/probe/test_sharder.py:TestContainerSharding.test_shrinking`\n * Run `find /srv/*/*/containers -name \u0027*.db\u0027` to see how many databases\n   are left on disk. There should be 15: 3 for the root container, 6 for\n   the two shards on the first pass, and another 6 for the two shards on\n   the second pass.\n * Edit container configs to decrease reclaim_age -- even 1 should be\n   fine.\n * Run `swift-init main start` to restart the servers.\n * Run `swift-init container-sharder once` to have the shards get marked\n   deleted.\n * Run `swift-init container-updater once` to ensure all containers have\n   reported.\n * Run `swift-init container-replicator once` to clean up the\n   containers.\n * Run `find /srv/*/*/containers -name \u0027*.db\u0027` again to verify no\n   containers remain on disk.\n\nCherry-Picked-From: Icba98f1c9e17e8ade3f0e1b9a23360cf5ab8c86b\nChange-Id: I7b7be3b62e033ab668316e3c698ba667ceef31f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef68d69c586f9231a20b24d973893b154353290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef68d69c586f9231a20b24d973893b154353290"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
