)]}'
{"id":"openstack%2Fswift~777284","triplet_id":"openstack%2Fswift~stable%2Fussuri~I324916f8207e1346806951b600105cccfd8c62f7","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"I324916f8207e1346806951b600105cccfd8c62f7","subject":"Address a sharder/replicator race","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-26 23:50:40.000000000","submitted":"2021-02-26 23:49:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777284-ussuri-sharding","meta_rev_id":"2ce69830243f26f8867757c797640fbeeb44d917","_number":777284,"virtual_id_number":777284,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 23:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 20:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-26 20:17:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-24 04:30:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e104ca50cb865c7ab3c76dae076508543e3f02f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc1bc33b6808169644df1a5080c4436afdf2d9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:30:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194189fd1fd64cbb8d260efff14b80ac : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/2c521a195db24246a5aabc51f85d12f6 : SUCCESS in 1h 09m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/819cee4d3a5948c49a73516ade535bdf : SUCCESS in 48m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b663776b50934eaaa3d667fa6b50cf92 : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5058f2f951fa43e08ed8215ea5221e56 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f975f57114c4dfe8f3c024444d26dfe : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d811219f32d432fb99f5a7216f6522d : SUCCESS in 15m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/420ca331cb524b23affbf0aeb43ee5a6 : SUCCESS in 12m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1982fb8cae4542abaa9de00959ed62a8 : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34d64b5499204222810ca4eb4e0b3c7b : SUCCESS in 11m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/98c0854ed1634893819074a382030439 : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c6a07497e2a54e1d8c8b537f6c949da6 : SUCCESS in 10m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/633f59dcd0a54fce8613059c8928d491 : SUCCESS in 10m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/34b0772122484903b0662533d78caf32 : SUCCESS in 14m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8df2329c8c8a4798bdc23d4781dc343f : SUCCESS in 10m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4c1988daa30948469d5c429b092068b9 : SUCCESS in 12m 22s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/38e2e0a050c846fbbe5d4fce15ba20ba : SUCCESS in 12m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2a7348d329d34f64a6b762ec8ad18b33 : SUCCESS in 13m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/81fea91f02ce4b56956c798366e5aa50 : SUCCESS in 15m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13c3077521be4c1a85dcf3264206f8e1 : SUCCESS in 19m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b9cfebfb74584ce1ae359343ffb00732 : SUCCESS in 42m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e27ad44392f34b72b408c129957769ef : SUCCESS in 31m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4e8b27b252d14f1285fcb7d2a405be0a : SUCCESS in 25m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a08f7f2b7b1a40d894b0783688b362a5 : SUCCESS in 28m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60081b1068004842b44fc09a99cf02d5 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9037c162cf9a4b4b8b416a02aa7c0186 : SUCCESS in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c0bcd47d0db47c7b08d039f447f3669 : FAILURE in 9m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be36405135724fab9d58b1ddf6fa8a15 : FAILURE in 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"60f355b2e27ad35aef790e5560cd4736266c5345","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 20:17:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, originally shipped in victoria.","accounts_in_message":[],"_revision_number":1},{"id":"7b4e3229986ae083ffc404beff1a453b9bee497e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:30:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad574459ef7454a8016b57ebf07718c : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/988e8b76019a4abd8f65563424f68f03 : SUCCESS in 1h 10m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b6d90cf2b864bf1828f0f3178613590 : FAILURE in 39m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae7a43ae7241487d8d46abd22a5e5bc2 : SUCCESS in 10m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/90375a77db6041ecb80abf142d1aff65 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40131e28c98a446883da0e77d91ae028 : SUCCESS in 24m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/96a08be8906a4f1b9eb6d3b776678085 : SUCCESS in 30m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35d58985ccf743578e3f3e6e63ded1fe : SUCCESS in 23m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/425919645490481e889b382bfc86994d : SUCCESS in 33m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1f065107eacc4e7e9debb781b3f5c709 : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0551f9d7f5d748d0a0d99b62a17ccb51 : SUCCESS in 13m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cc70dfd5d4b44b359c2d1bf7b3798b75 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ce3374b0cb74644abb7ca1dcf68f91e : SUCCESS in 17m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1ce470bd12ef458c9c06f0bce4e3bc40 : SUCCESS in 29m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/99bc1d294b5646da90f9e380aa6faf57 : FAILURE in 29m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/19518b67d3094d1998feb32a8e416acd : SUCCESS in 25m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f777cd1134c547c5bbc33b1c481b63ca : SUCCESS in 17m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9c8e6c0875054063baf08b1b9a813364 : SUCCESS in 23m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/55a0a9e9d1ca43c084c5fdd0775c392f : TIMED_OUT in 40m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59b4f6e306e543ad8901ce5930777b34 : SUCCESS in 43m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/74b201f5e3ee42f19e4c74f2f51ffe9e : SUCCESS in 44m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77a970e6ce814636b38e1f75cc060e7b : SUCCESS in 49m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b18cc8736a12428ea38a8940618346e6 : SUCCESS in 49m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e3cef7df6114a95a7e52d08c00f5af4 : SUCCESS in 46m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2622892df4c04d579d38d5ac781aecba : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15aaf054b03142b8a3de587ddb5f2f10 : SUCCESS in 7m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/652ba4dc859b44319d92c04815de0248 : FAILURE in 10m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/519cf2c75d26468fb55a7a8a71cd30ce : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1833b6d42758adfb903fcd3ac5693f13588cb7e2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 21:36:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ad9b3442e50c8d30e3fa529b3c737f1ae1d0c879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:36:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7265a99e3f4db9a56ec5d267dec8fe : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9048b567577346329a77c392cdf61ce6 : SUCCESS in 57m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2b5c65646314ab6b83e8a0346a83d22 : SUCCESS in 43m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f764a9cef64947ccbfdc844bfcb6dc5e : SUCCESS in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2196d3cc441d4985bcf5d4be3782b372 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d89039ee0a5246e688c1da046ea4a5d6 : SUCCESS in 12m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ab12e91b1d04e9d9786a194296da451 : SUCCESS in 12m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc0e15e62fe7455689d449a08ba5b6c7 : SUCCESS in 21m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/de2e8aafaf2a4bf3b5de1a01cd574fd1 : SUCCESS in 20m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e0c8940f8f64fb4a8ac8de67dca3451 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f28d47231e55453ca7721195b132b3b0 : SUCCESS in 21m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/03542998bea24c879969122b938f9c8c : SUCCESS in 9m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ffbfba78255435ca0ebae923bbe2318 : SUCCESS in 19m 31s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/715c89adfc4843518fde292d05015dee : SUCCESS in 14m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/47d63a417808444897b0b153c62ca5d7 : SUCCESS in 10m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6465264deea74fcf8dbafd5a2454f63a : SUCCESS in 12m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/144f8b4b8a9944e495c33f477a9be43b : SUCCESS in 9m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8e76053a6a94450ea2c0ce30b9b63ad2 : SUCCESS in 20m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b78a7e90ea60499f890892b137c6e8f2 : SUCCESS in 16m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a6a44f905dae494c89436abe72320fa2 : SUCCESS in 39m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f2223f3abe0544138450501e6bc1cbaf : SUCCESS in 45m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0468f72ae2fd4450a50b8b5523247f89 : SUCCESS in 25m 47s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7fa8667e536d4baf900e979cd6dd592c : SUCCESS in 25m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a0205cfba724ddea66acced998f2a54 : SUCCESS in 28m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c26d6fd8084c4a93970024bbe1743a9b : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb20cd823b6b438db3b18c2406ac7599 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/961fa564ec9648b39999e4ab6280c258 : FAILURE in 10m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d706d35a358244af9d8b04f09bd761de : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":1},{"id":"dc5899efcefbae600473cf2e5a2239929bdecc8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:36:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"754622911c1bdbebb9c863ece4654dc1464d6e9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:49:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5451169b74864d969e676606d5a983d9 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/d14c63257a61421fb77acaa54e2df58c : SUCCESS in 1h 11m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bab411c07f034a1cb1f4e6f3d5a98c16 : SUCCESS in 35m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c81820ca98fe4ab78a4baab2ea0d749d : SUCCESS in 10m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c87f7b1194de4f148964f9486da41462 : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/670fcfd8675c4f259899d781054a2074 : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/762067ee358a47549a4136b47c8bb615 : SUCCESS in 19m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afe886ff32da4c63af65195dd251c799 : SUCCESS in 21m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42acf4a04372440fa4eb2825541f0328 : SUCCESS in 21m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb935cb26d4b48fabf8bac625f60ec4a : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6074c960a844da39bd09b9898d32bf8 : SUCCESS in 13m 38s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5141e433ecbb494cab66be9f0ee8e664 : SUCCESS in 21m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37b928faa3224ccea3281857d8d6dba3 : SUCCESS in 11m 36s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9e3c463d7e7c4e96bf06ccd2bf087b09 : SUCCESS in 31m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/41df8c8ccc71417db07f16da54567868 : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cde559ea36b54727bb6a0d0c90fff11e : SUCCESS in 22m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f349c8edde314327b6ae09b4105d1098 : SUCCESS in 14m 47s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/60ac401b55a64d62921a75748b61be76 : SUCCESS in 31m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9824a3375c0d4f58bdf4213a7e4f067c : SUCCESS in 51m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2f11d3884354c329ff395f53d367e08 : SUCCESS in 42m 38s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5f006772bb19457392bff29f04a6dc5a : SUCCESS in 34m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbdedc0bc9904d38af8b5dc2be092482 : SUCCESS in 41m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/175a21f94463463e92343621d4360c01 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13d452c41654323bd65fd9abf96b83a : SUCCESS in 8m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12136f26beca4d72b533670d7ef498de : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3c97de7523dd10c2c354cccec3a9bce00efa9b43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:49:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2ce69830243f26f8867757c797640fbeeb44d917","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:50:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784c4994cded4173b41fa9c311e5fa9c : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c4b622d7e2d948b2b38777c9cae9a261 : SUCCESS in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/35335356fe74442a9883b2c562fc5063 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94346b1345b28fa1809bc0def34f821ab13ea0e4","revisions":{"94346b1345b28fa1809bc0def34f821ab13ea0e4":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/84/777284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/777284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/777284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/777284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/777284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/777284/1"}}},"commit":{"parents":[{"commit":"eef68d69c586f9231a20b24d973893b154353290","subject":"Identify deleted shards as shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef68d69c586f9231a20b24d973893b154353290"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-25 21:08:34.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-24 01:59:30.000000000","tz":-360},"subject":"Address a sharder/replicator race","message":"Address a sharder/replicator race\n\nIn the previous patch, we could clean up all container DBs, but only if\nthe daemons went in a specific order (which cannot be guaranteed in a\nproduction system).\n\nOnce a reclaim age passes, there\u0027s a race: If the container-replicator\nprocesses the root container before the container-sharder processes the\nshards, the deleted shards would get reaped from the root so they won\u0027t\nbe available for the sharder. The shard containers then hang around\nindefinitely.\n\nNow, be willing to mark shard DBs as deleted even when we can\u0027t find our\nown shard range in the root. Fortunately, the shard already knows that\nits range has been deleted; we don\u0027t need to get that info from the root.\n\nCherry-Picked-From: If08bccf753490157f27c95b4038f3dd33d3d7f8c\nRelated-Change: Icba98f1c9e17e8ade3f0e1b9a23360cf5ab8c86b\nChange-Id: I324916f8207e1346806951b600105cccfd8c62f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94346b1345b28fa1809bc0def34f821ab13ea0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94346b1345b28fa1809bc0def34f821ab13ea0e4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
