)]}'
{"id":"openstack%2Fswift~777285","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ibf304bb64f005a86cb4d08312c5d7ad4a2488719","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"Ibf304bb64f005a86cb4d08312c5d7ad4a2488719","subject":"container-sharding: Stable overlap order","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-27 09:52:52.000000000","submitted":"2021-02-27 09:51:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777285-ussuri-sharding","meta_rev_id":"bb8d7069d46a6c97319044d3b6e456b198edd12f","_number":777285,"virtual_id_number":777285,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-27 09:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 20:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-26 20:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-02-24 04:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 20:18:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"8b976573f2fa9146de273985f86bb8c53574b20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25e2607f769d930033d1d9fe6253702917ca79a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:27:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f29e07b8b5421787c106efc1787fbb : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/823610d9f12b4da383c98c8f90821d3f : SUCCESS in 1h 05m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b290ecda74343ec97e75d69d605d0e4 : SUCCESS in 36m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/003f228f09a048d99595713dffd4132c : SUCCESS in 9m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f359684b4ce4a4db647dc4f9cd7b23f : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9e47b6945c604b7f870aa05236cb39f9 : SUCCESS in 13m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a731a61533a4081a9c5fa8a5a3199a7 : SUCCESS in 13m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a4224f2d89d41a5b2136f7e817ec7a0 : SUCCESS in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a44224d9dcca4128b77857b93cd007f5 : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae5fe58a7c9943bc87931006605e9549 : SUCCESS in 9m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20b88b1fcb77436285e9cb28055f4090 : SUCCESS in 12m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/63564f26fb9d4629ad3ab11d8178119c : SUCCESS in 23m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9041f71ffd4747589bf63ecba30a542b : SUCCESS in 11m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/21524325600e445e999a41084fc766b5 : SUCCESS in 14m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2b50af5153cb4d3eb579463e8fb635cb : SUCCESS in 10m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8e15c30b494141e68cd1f8c8f9164808 : SUCCESS in 13m 05s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3c80f53685fb4789946e090295fe2fb7 : SUCCESS in 10m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/003e44b82634499b927798d67d24872f : SUCCESS in 11m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4c69058f33a04d7fa51be98488e40151 : SUCCESS in 16m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c93e2fdd92144dacb5926e418650614d : SUCCESS in 19m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/601cd896318748268e1a26d427ee90f4 : SUCCESS in 42m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/88dc61ef18f24a8ab8ca5b3b02ce57cd : SUCCESS in 31m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/29220a8dea204bc79655e9a1bba7aa93 : SUCCESS in 28m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f5acf708ca14150a7e2093fe494cc9e : SUCCESS in 28m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be2d0f65ff8f42de9b4b4b94c25f0f35 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90ec833eee3943f79cabb9491bfde9d3 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4a67c71cc87491c84491601ddce4eac : FAILURE in 12m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05d21b5801504a10b5faa41044aeb5ae : FAILURE in 53m 35s","accounts_in_message":[],"_revision_number":1},{"id":"729953320414bb37571f75de3559651514317bf9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 20:18:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b06a2a5f37909b60d914e88cdf67ce46287ceff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:21:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7925b7e2dd5d481ea2e26b1b5078db3a : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/3ee042be298b47f6a38e87ee4ed737c6 : SUCCESS in 56m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53b906651c984e0c9a5caea117d620cb : SUCCESS in 36m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/246bc66186d04ef98251a5a035cf41d5 : SUCCESS in 14m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/40bd6147b83d45b9aa36354abaee7e75 : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51e3c4e8ddef4cdca115333972f55a3b : SUCCESS in 17m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf57d81ec4c48a9b3e6ecd5eb08defc : SUCCESS in 13m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed321b8397334777b65f49dbafeb0bfb : SUCCESS in 14m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8651d4de0834302818964a392e1e710 : SUCCESS in 14m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ceabfcefaee74d21b9f256fc31457124 : SUCCESS in 10m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/452f7b926b7d457f96055a50b81911de : SUCCESS in 13m 27s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4e9ae3c6f8b74dea9600ad5b9184bba1 : SUCCESS in 10m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d2a856fba0e845bdbd9ddf2675ecd376 : SUCCESS in 11m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/680b512dace14528a6fbc051e01b68ff : SUCCESS in 16m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b5befc61b5e8437e8906701bd135063a : SUCCESS in 11m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bfb749b838794ed4840741e9543c876d : SUCCESS in 13m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3569186c9e884f0a9b508cf3ce2cf9b2 : SUCCESS in 12m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/41bf8664912c45589352f5b5185d768e : SUCCESS in 12m 15s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/445298ab01b948b5ae4468d764c607b5 : SUCCESS in 19m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa3bccc91dfd424587bbd577245e8842 : SUCCESS in 21m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9d52ba1021ee48e8a2db01f7e95ebcc4 : SUCCESS in 46m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5857790775ac4ce6ad06d0867681e112 : SUCCESS in 29m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9f2cd490426b4d698d2fa646dc24def7 : SUCCESS in 31m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/894d1f14675444f0a76ff3407fd1eb50 : SUCCESS in 32m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c560585ce004551a7a80f016970b651 : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a1adac5545477cbe579a05c2186c4b : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5761bf08935742f1a786f80bf051c733 : FAILURE in 10m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e73376a0283640748b5c41408aba2da8 : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a917ef36a6bfbe477f4005e124f423c8a346955e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:36:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7476264c4576ed82fd45b49f7ccb29aa291384f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:59:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e1e172aa3746e08489a236a035e5fa : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/1269d1c661ca4ff5b3638f4500d6a941 : SUCCESS in 1h 07m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e678946ee054563a150e07d2e79fb77 : SUCCESS in 52m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/587343e014804c7a8aa126826f7eaffb : SUCCESS in 12m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/59c7f8f320a84176ac0a454dee4847ae : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1fcc0a68edc14f78b394721706db0419 : SUCCESS in 23m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/78707d138ab14955b684cc31bf8cbd44 : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cd0586dd573e4156bea200c41443ea73 : TIMED_OUT in 40m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bc8c716e832647ed8a1c1d029c7b87f1 : TIMED_OUT in 40m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1f97dbc7617463a875149299edcab84 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0f3ed9bc0674d2189f636f16512a4a3 : SUCCESS in 19m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e3da9e513fec4429a74af4798d86b17c : SUCCESS in 28m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c56e586edda74fc49962c02d0f0aaf9b : SUCCESS in 17m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d1337215a14e4b9db2bc56a3fc7474fe : TIMED_OUT in 40m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b49b83d89b654cdc82db407b2ea17792 : SUCCESS in 9m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/859002fa72a84f3b8f3e83c091574b5f : SUCCESS in 11m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5fc7d82cf0ef41588720696c3b25c87e : SUCCESS in 12m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ff1b6662c5dc4b5ead0c3709382da25f : TIMED_OUT in 41m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b9fdd36b6bac4c35908dcce65fd1a831 : FAILURE in 38m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e117640166046a7ad8d97590b880f4c : SUCCESS in 27m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6613ec80373b4929837a0386918e307f : SUCCESS in 42m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91acedf32f6b47ec9a58344774f9c1a2 : SUCCESS in 31m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3efd53950a64d828b64332dd1e4363f : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82ede932d6e4f19a9e5e233c2724d9b : SUCCESS in 7m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c6ae6905d448e2bc2febabf0896a66 : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7b8e1f8cb84841259ca7a42c59549b248e25b7c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-27 05:38:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b80526ead55aa9f8d7c7d9a3e2d19a3be2663c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 07:03:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9722402b624b31b7e62913cfe44877 : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b5da2c4d28b741cf8652c49e21a939a0 : SUCCESS in 53m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/019a0b2083f24e75b1ebe29b00a17fbb : FAILURE in 1h 14m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31654906013544c6a9f1e6e14c31b555 : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cd8a6b2bdfda4543a3fef76f85959376 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c8166bf92814e1399746e8872c2951f : SUCCESS in 29m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c43e393f9ade4e9cb70aa14c64252837 : SUCCESS in 25m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddda8ff2cab34dd599c70cf8b94829ed : SUCCESS in 40m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/25792bfb68db41f3a4b740f0f1241a41 : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12180b2d0af045eb86e3c6e263f7bfc4 : SUCCESS in 9m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6a3c466ed12440d8c9d6d42fa6a1352 : TIMED_OUT in 41m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3989eb24525348f1a82c05aa3cfd4d91 : SUCCESS in 10m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/19e46b46e5104da09408be14e0b24c79 : SUCCESS in 16m 06s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/49367ef9703c44afb0221c155364ae63 : SUCCESS in 20m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b9b174c4eaab423ca632d61b29bda6b9 : SUCCESS in 40m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/00cb2cfdd01049299d8b8e7473e54662 : SUCCESS in 27m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/21e99a96342a4f6a93ec5e67294600aa : SUCCESS in 11m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0bd1eb8780214161a39e7ee931ca63d4 : SUCCESS in 32m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f1d2d574195d42799add5c191505b006 : SUCCESS in 29m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a124a4c29fd4862baa3b97d4dbc0b45 : SUCCESS in 40m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f24feb82fd9d4a40aa52cf31437cdf44 : SUCCESS in 1h 22m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fca517e7b0fc473c9777d8e6fd12a3e7 : SUCCESS in 32m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4b42c36376ba4bc0be876ec9175d748c : SUCCESS in 40m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7edfc1b4a2174a6db96f4d8753f6480d : SUCCESS in 33m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51abdb2e9754400ab7e152b2dea40023 : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74b31eda0a7f40d7857f581cebb0196e : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8075152821ac4d4a9e5fa8adece9ef40 : FAILURE in 9m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9075d001ff824804bec24e92022ae6ad : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a5574ae3412907170598631ae4938e36844bf853","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-27 07:39:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8a58c2d8b058ce55368cf67a5fae5b370bdd12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 08:46:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b5b073f4514129a189b4586d6ca001 : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/05b07739c1904b8e817508cf5ce03590 : SUCCESS in 49m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/541e1d65657546e58876d825d232c4a9 : SUCCESS in 56m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f5abcd4216294e49983fb6f770b0b1b4 : SUCCESS in 10m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/974f335b3a3a4d4883a06c39fe3aaa32 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b89c7db029c4f42b947aeaf32d4723f : SUCCESS in 14m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29157929583a43f5960e1b3d33ef9f6b : SUCCESS in 31m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d7f0434e00514049969e3f4c65fb356f : SUCCESS in 24m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/95ccd404d8574b44bea729c83b60902b : SUCCESS in 35m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e74ff15085da4932bbe7134a8b5062fc : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fbc33b6522484a32a2b63e79c594f19a : SUCCESS in 19m 32s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/62abf2fd2a1546f99facd44019a138b4 : SUCCESS in 21m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0eb431c76157450eb39a9402be634b4a : SUCCESS in 12m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c73c6c7dbf5f48c5acd8986a1c0a66e1 : SUCCESS in 30m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/775e22c16ff0401a871237b7bf8927e7 : SUCCESS in 21m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/81d00ad7fb4f4112a9e228fbe42e62cb : SUCCESS in 14m 00s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/75049e6c8a4e415386f3c92295bf2d03 : SUCCESS in 21m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3a940a9166134d5ba01ba8ef11990a97 : SUCCESS in 30m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cc226b6df1a0465bb448082ae4539e97 : SUCCESS in 38m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2060ea0639494ed3917b447f80beed3c : SUCCESS in 23m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df9a336f74b24e4dafa1d3aba46797d1 : SUCCESS in 42m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ce065fc791a4780838e5cc65366c276 : SUCCESS in 42m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e3019f2381564415a05a155e9e27146a : SUCCESS in 40m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a60cfa10454446f98cf31890e4b07a39 : SUCCESS in 36m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/398fa23625cb47a69a2a0863b070fb7b : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e506bb283b64f48b434c3a2304c686c : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96b2e3b972a243b8b9fb74827c9041be : FAILURE in 11m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bb3f1947a484832b8449ff66d29eb19 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":1},{"id":"da07137fd95d115b51257428449e5812dce260a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 08:46:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"890d445addedf8cfca488a94dbe586cf3e631289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 09:51:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf62164792046f895eebd58c0e6614d : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/f49d4d44e4f340e98da33aaac70d5a91 : SUCCESS in 54m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a62d8d8104641bfa3f980c8e20a7fb8 : SUCCESS in 35m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5900a79dda8a4a599f3dee86450998b9 : SUCCESS in 9m 49s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/af4947e73b19487a86a70593e63a4033 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77dd3a4a972c482c8399f416d97fcdc2 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f54b9e2b5264591b39f7749e8f4d8f3 : SUCCESS in 12m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d785e51538f740c999e0a3ce29b08ebf : SUCCESS in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cba9b00caf4f4362b7ea2454e8c7ff5d : SUCCESS in 13m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cc36366f5854625b4803ff4598f9503 : SUCCESS in 9m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f50a11e864644d9da9136eff6dddc4fb : SUCCESS in 12m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f4448d14386a4dcb8e1edc20e598c741 : SUCCESS in 11m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05369b50cb75403b826fdb95ed4146d5 : SUCCESS in 10m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d5bb9acf5c4d48cd97197283d2a5d3f6 : SUCCESS in 14m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dfc3714fb07047a684c5220a7f5c9507 : SUCCESS in 11m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b2104c6c450746a7b430461266aac027 : SUCCESS in 10m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/892b5243d9284ef6801e26ac85812d22 : SUCCESS in 11m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/55e3d58b2a3e490e80582af4e4e81dd2 : SUCCESS in 14m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5fa4e74a48e64a799c01f876e617b1d2 : SUCCESS in 46m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/35413ada3f6248d59c48b3ed13f40181 : SUCCESS in 24m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/17c106d32a5042df869bb8932e62934c : SUCCESS in 23m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6957e160753e4dc1b20edc3feb97fd17 : SUCCESS in 32m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0bab5ab603b49c49f4f01e572e35ef5 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94261af6a9a4e3b84aeb6f3b7dd434f : SUCCESS in 5m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c45ed348db1e4ea6968d95ff9de17d9f : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8f5f42b631c7856f18c0e54c309257c393f80301","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 09:51:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bb8d7069d46a6c97319044d3b6e456b198edd12f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 09:52:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef869758c4874c73ba7fb7cec7038c0f : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d0e1d63d1a3d46869c65f01a4d47a6e7 : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b7bc034daff348e8b9ff269f5a834e73 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"760528cb70cf56a1a23a7d3c3e3b708677213687","revisions":{"760528cb70cf56a1a23a7d3c3e3b708677213687":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/85/777285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/777285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/777285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/777285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/777285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/777285/1"}}},"commit":{"parents":[{"commit":"94346b1345b28fa1809bc0def34f821ab13ea0e4","subject":"Address a sharder/replicator race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94346b1345b28fa1809bc0def34f821ab13ea0e4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-22 20:17:08.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-24 02:15:13.000000000","tz":-360},"subject":"container-sharding: Stable overlap order","message":"container-sharding: Stable overlap order\n\nWhen you have overlapping active shard ranges updates will get sent to\n\"the first\" database; but when the proxy queries shards for listings\nthey get stitched together end-to-end with markers.\n\nThis means mostly the second shard range is ignored.  But since the\norder of shard ranges is not stable (it seems to depend on the database\nindexes; which can change when rows are added or removed) you could send\nupdates to \"the wrong\" shard.\n\nUsing a stable order leads to more correct and robust behavior under\nfailure; and is also better for cognitive overhead.\n\nCherry-Picked-From: Ia9d29822bf07757fc1cf58ded90b49f12b7b2c24\nChange-Id: Ibf304bb64f005a86cb4d08312c5d7ad4a2488719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760528cb70cf56a1a23a7d3c3e3b708677213687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760528cb70cf56a1a23a7d3c3e3b708677213687"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
