)]}'
{"id":"openstack%2Fswift~777286","triplet_id":"openstack%2Fswift~stable%2Fussuri~I12988d2cd8e7c948f107e7df943cab8f8565c3d2","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"I12988d2cd8e7c948f107e7df943cab8f8565c3d2","subject":"Refactor audit shard tests","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-28 02:57:16.000000000","submitted":"2021-02-28 02:55:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777286-ussuri-sharding","meta_rev_id":"ce26a7165c9fa12fccfde22b391b7cabed1c7197","_number":777286,"virtual_id_number":777286,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-28 02:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 20:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-26 20:22:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-02-24 03:59:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 20:22:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"ca920e9cd47dfbbfb68b1574c1692b3c78242e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff6d97f6b7779b73dd7d045dfba57a503bbab530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 03:59:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07eaa5250fe547a09e51df990794210d : SUCCESS in 10m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77042a9dad974704aa1c4a3860df3938 : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/86767014f87542a7ac73fdf785b3af19 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/310c326dd46d4f359f543fabe6f04be7 : SUCCESS in 12m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34ba6dacf26d446b857776f63ae3c934 : SUCCESS in 13m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8f6cbf1de5254f4882fb9464880183d5 : SUCCESS in 14m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2af96bb6a7c84f0e9a67db4c9d8d25a9 : SUCCESS in 12m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66ea708b5c7f4976a14f007606a8ba71 : SUCCESS in 9m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f57caa72abc2427cb9302614450af5c7 : SUCCESS in 13m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1ad9e451cd8a47deb0b4f8437693fd21 : SUCCESS in 9m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/288ea898bea14d73bf54ae1269fbc6f0 : SUCCESS in 9m 59s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e81309f99c874f2fa801ce994874c95c : SUCCESS in 14m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4a6b8d7e238949c38f9bbc84ad3da911 : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b77f36014faf4b11aa1be013c451b8b3 : SUCCESS in 15m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/96f4c917f229488ba13df7f67b4c0dec : SUCCESS in 10m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cb136905012c46579421e71486b47d16 : SUCCESS in 11m 13s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/edbaca100e6b45d794431abf5f139c14 : FAILURE in 15m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3bc2a77aa8c4683ad58c64bff269aa7 : SUCCESS in 29m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c5c739cd64284fc88f24279325555748 : SUCCESS in 27m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d9cf9c846fa44aaa47878ffa618df6f : SUCCESS in 26m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0fdb559dd5a4126884c70882e99336a : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/098a579e6c7e4871be188c9399f37ead : SUCCESS in 7m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/907a94ca10884e11a94396fffb8f3bfe : FAILURE in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1acd5bede0e81c22eb3637b322837e9b19b172d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 20:22:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nWill ship in wallaby, but already backported to victoria as https://review.opendev.org/c/openstack/swift/+/777033. Clean cherry-pick, tests only.","accounts_in_message":[],"_revision_number":1},{"id":"97c516f836d13a91d17e83994fe943e4ea74222f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 20:53:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e2391ab3a84a21bc01d392d228324e : SUCCESS in 9m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bff0d909bc3435691ccf80ea474ee75 : RETRY_LIMIT in 2m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24ece9df762644bab19a8c1d798b733b : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e02ff31cfa4f4cc4bfa077d4d651d015 : SUCCESS in 14m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb5296f35ff04040946d0234784dca5f : SUCCESS in 13m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/028de22c0283432287d01c91cf0654e3 : SUCCESS in 14m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e2e03b9dfc234cefaa28dcdd023d2f5f : SUCCESS in 13m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e53bad0f1ed43d0ba0f9d00bc74317f : SUCCESS in 12m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dacdf6c26957424a8e96c9bddcf82e46 : SUCCESS in 13m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f09773f4abea41a99431df285dc16094 : SUCCESS in 12m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c79c88971fe34d9785d2b3ad9c1edf76 : SUCCESS in 10m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ea4b496948bd4f3abe989032767a333e : SUCCESS in 16m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8a5d5699dfcd411c8daabd0d1b772714 : SUCCESS in 10m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/828fd2ffea7e420780862790e02270d2 : SUCCESS in 16m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/658a33b76c704ab79b5c61069b79094d : SUCCESS in 11m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8187bd7c60f5409c8c1d0e46de4d495d : SUCCESS in 11m 07s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/fabac0326f3749fbb05564950dba2c2c : SUCCESS in 17m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22a6067ad507401cbf5a9dcc849b17ae : SUCCESS in 28m 19s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/14f4a22f30d8441ea98c5d62d419aaa8 : SUCCESS in 28m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40ae8d60545b4aecaa34d946ded6c7da : SUCCESS in 29m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b43997794db4080bbd1d5218bd17cee : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1eb332ffa7c423db946bea7ef34d5aa : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc139b461c5846008826c9be951610c5 : FAILURE in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90d0d201323c6f6697032d530a0cd17bcca5d8ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:36:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02507b47f0a4371950987ea435df0530a7b30266","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:59:42.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"7e2773ca78dc7b803e3f9ce7349fff4be29a40a4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-28 01:22:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62f39d0639db8662b5ad537f54cc8ed8d3eeb80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:06:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12155fc5c264f29835c7208e94d2278 : SUCCESS in 8m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91a0ce5bd0964c9ebf907be746f06457 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a315c4d014c475899e5613f3b6d3177 : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dae9e4ef2484475e85fa24d073f2eb90 : SUCCESS in 14m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7d63c23f24d4a71b63e28377c575754 : SUCCESS in 13m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c9582115f18746379118060b61d24732 : SUCCESS in 13m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/10cebdbe07ac4b109eeb45d58241c167 : SUCCESS in 14m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7ac8926b6b1d49cd823059144b2139f4 : SUCCESS in 10m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45803888269749dd8950ffc9c98a158b : SUCCESS in 12m 52s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bbabc5f70ac3424ea85e59f78a97271a : SUCCESS in 10m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12e257aeb6ed496390dc16915e2cf634 : SUCCESS in 11m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/96e84703caea4752939f10bec7d99c18 : SUCCESS in 15m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/da9e9677f7d8419abbe8a7ee229fcca6 : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/00fb8c3d65f64f2c88fd4fd1a71c6829 : SUCCESS in 13m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8f4889ce775b40c9be1fc54f24ec1d57 : SUCCESS in 10m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/afdd19a400cb46e082d899ab17f389cb : SUCCESS in 11m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/47e14d17ae064385a911c6d5c465fafc : SUCCESS in 17m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1f177e270c245fc8de858299bbfeb6b : SUCCESS in 30m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9561e97e7a0646c898ec30692bc454d6 : SUCCESS in 35m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82f6efca977d40ea91ba448255183faa : SUCCESS in 31m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c06fa9c6c094a66b742a5501fd47a9a : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b739b5230d684884be253f35125760cd : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a6a137fa1e1f442db3f4fcf390a03784 : FAILURE in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83932c23ceded027e9064564571f77e714d84f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:06:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8dbe3768a431b6f5932b52c5c9cfbaef89d826f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:55:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bb21a19cb3499f96b42c91870e3f22 : SUCCESS in 10m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3f521a8061884d19b368f0db867c399e : SUCCESS in 8m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ca216012f63c4112b9692306acce4710 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8ed8209fc2a4961a7335c004c24722b : SUCCESS in 13m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1921037bf4ea4e25b516f3fbe37d7046 : SUCCESS in 19m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/53b5c95cd63344d782710a80adaeb529 : SUCCESS in 19m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/15c459a8d9f8423fa7a3bc5582b9b261 : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b38caffe621942ecbe6bdf691178a120 : SUCCESS in 11m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c5b526d7d07441abc45fc3891f52c56 : SUCCESS in 12m 44s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a05388c6c84e46fc84f43cd8c45c32e9 : SUCCESS in 8m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/58bbcee928984707b279790ed36801be : SUCCESS in 14m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ff8e2c5fd1fd4bf187a9d88de700ae41 : SUCCESS in 14m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/68a54337e7fd4d6d88486e2f04b975ae : SUCCESS in 11m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d97e480992324f1c826a28a9574e1732 : SUCCESS in 13m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/769151f8441f409ca373d8cb9bff1114 : SUCCESS in 11m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f2cc8885d6044b828dfeedda8a3ff7d3 : SUCCESS in 23m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a436a0c14704e70bd71647c947d54d3 : SUCCESS in 30m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/448240806e214315846625fa6a6bdab1 : SUCCESS in 36m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/28faa03eefd946d187e069c9ccf358b5 : SUCCESS in 31m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10d876fbc5b04a07b372259a422e232f : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d7cc064fea4c34b4f70e4789f8fff6 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3b2bc84732ea9f04638e28e94c045b7c55b21c40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:55:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ce26a7165c9fa12fccfde22b391b7cabed1c7197","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:57:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e7b22a0ba342a6883ce9f6f7b14c7e : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/65aa921e375743518354d73e67ea50c1 : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/babcb236907a4ef9873554a419653663 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"959a98d7b18aba3e9b9812f5cf81d107a237e59e","revisions":{"959a98d7b18aba3e9b9812f5cf81d107a237e59e":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/86/777286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/777286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/777286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/777286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/777286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/777286/1"}}},"commit":{"parents":[{"commit":"760528cb70cf56a1a23a7d3c3e3b708677213687","subject":"container-sharding: Stable overlap order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760528cb70cf56a1a23a7d3c3e3b708677213687"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-03 16:40:13.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-24 02:18:42.000000000","tz":-360},"subject":"Refactor audit shard tests","message":"Refactor audit shard tests\n\nThe existing tests cover a lot of behaviors and carry around a lot of\nstate that makes them hard to extend in a descriptive mannor to cover\nnew or changed behaviors.\n\nCherry-Picked-From: Ie52932d8d4a66b11c295d5568aa3a60895b84f3b\nChange-Id: I12988d2cd8e7c948f107e7df943cab8f8565c3d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/959a98d7b18aba3e9b9812f5cf81d107a237e59e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/959a98d7b18aba3e9b9812f5cf81d107a237e59e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
