)]}'
{"id":"openstack%2Fswift~777287","triplet_id":"openstack%2Fswift~stable%2Fussuri~Icdf1f9d02d1e6ae5dc1112d2792ef6845067158b","project":"openstack/swift","branch":"stable/ussuri","topic":"ussuri-sharding","hashtags":[],"change_id":"Icdf1f9d02d1e6ae5dc1112d2792ef6845067158b","subject":"Use swiftclient Connections in sharding probe tests","status":"MERGED","created":"2021-02-24 03:15:09.000000000","updated":"2021-02-28 02:58:18.000000000","submitted":"2021-02-28 02:56:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777287-ussuri-sharding","meta_rev_id":"9ab3b2f5755e0180bab975c1a9e37e56ebf91643","_number":777287,"virtual_id_number":777287,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-28 02:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-26 20:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-26 20:24:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 03:15:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-02-24 05:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1adb98f63ac6845982afa44177a3ed03367d7f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 03:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e7b0703ff49e4026cfa0095eabfa6b22f617586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 05:29:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/addadaf33db440cb85be5d025a583e07 : SUCCESS in 8m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c49de2523de1438684f57762f0064d7b : TIMED_OUT in 2h 00m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1129e8efdfd04237b6837f56a758358b : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2e6f71f267bfb7adbbbdf4856e99e043033e9f98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 20:24:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nWill ship in wallaby, but already backported to victoria as https://review.opendev.org/c/openstack/swift/+/777034. Clean cherry-pick, tests only.","accounts_in_message":[],"_revision_number":1},{"id":"467efcab32fdd5c499fd5e98aa6eeb2a83487bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:11:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86d46685f8cd4d238b4ca6f0b123ed58 : SUCCESS in 8m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c063017a6a3f4fbfadf7ab41926be86b : SUCCESS in 45m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365495e894da48f3bedd8eef955c84ee : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f65b451295f77858b96fe92a58de9fba62dd2911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:36:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"103e6535bfbb9df4b7cb00a5d53bb1c15104ff3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:59:43.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"771c77055308ba384837d5e3cab79c6a2840bb86","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-28 01:22:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18acfa27fe6fcd90a123188ce53a065f13304385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:16:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5cb3d4dd6d421b94a1dfcbc690d5e3 : SUCCESS in 8m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f3b30cfc1a549cea1e01734f33c092c : SUCCESS in 48m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf3f6ef5536469689a2f82728520bf3 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4691190a687d68ff21144ef5be1a8a1a60b04173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:16:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"679b69e748a92cd3e209e593329bafb0a3bb2fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:56:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613da2954bdd4cb69cb5851933674d12 : SUCCESS in 10m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6ba148a3575a4ccbbfa004ece059bca4 : SUCCESS in 10m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/750cb1ff2fc84916b85e15df6282ebdc : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/436148b665cd41f2a6909c489ccef472 : SUCCESS in 23m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/389797f73dcc43abac1a6c69a736279b : SUCCESS in 24m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63fe200c898a4f70a1947b170284c1d8 : SUCCESS in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/46a9e68008a445788ee93426caf010f2 : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/512794a474d848dab209cb288c445e47 : SUCCESS in 9m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6637e684e2a24a7c85309630f1764e19 : SUCCESS in 13m 40s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b159b181dad44f849953699bfc8b3d0d : SUCCESS in 13m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8683581efad548d1a46767974354316f : SUCCESS in 11m 27s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/04fe300c92e2454ba29d73ff8088b68f : SUCCESS in 14m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/af3fca03b8c04cd4b8cb7ed8bc734208 : SUCCESS in 14m 00s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e271f4bc803b43d2a0c99f868d61abb1 : SUCCESS in 10m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4c487556d000465bb15ba0c89f84c1ca : SUCCESS in 11m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/feedbe6954de46db8eea870191bba4bd : SUCCESS in 30m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fad761b6f7084cea874b7207a019b21c : SUCCESS in 38m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10ab813af584ee0972f8741b1c1ac3a : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7bac6baa9c6d73833ecef9e4e4e880f9ee39ef6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:56:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9ab3b2f5755e0180bab975c1a9e37e56ebf91643","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 02:58:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33113b97bc846c69da5af9313f18921 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fb6cb14edb86439a897831f7863b7afc : SUCCESS in 41s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a6102a9256f248efa571f63da59d4d26 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5791cec31e6935a9af691dbd181bc1657d87ce2d","revisions":{"5791cec31e6935a9af691dbd181bc1657d87ce2d":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:15:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/87/777287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/777287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/777287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/777287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/777287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/777287/1"}}},"commit":{"parents":[{"commit":"959a98d7b18aba3e9b9812f5cf81d107a237e59e","subject":"Refactor audit shard tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/959a98d7b18aba3e9b9812f5cf81d107a237e59e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-11-09 18:55:15.000000000","tz":-480},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-02-24 02:39:47.000000000","tz":-360},"subject":"Use swiftclient Connections in sharding probe tests","message":"Use swiftclient Connections in sharding probe tests\n\nThis gets us retries \"for free\" and should reduce gate flakiness.\n\nCherry-Picked-From: Ia2e4c94f246230a3e25e4557b4b2c1a3a67df756\nChange-Id: Icdf1f9d02d1e6ae5dc1112d2792ef6845067158b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5791cec31e6935a9af691dbd181bc1657d87ce2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5791cec31e6935a9af691dbd181bc1657d87ce2d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
