)]}'
{"id":"openstack%2Fswift~777358","triplet_id":"openstack%2Fswift~master~I30ead8b72cb616d1311ffc81d9cfecb1afc9a05e","project":"openstack/swift","branch":"master","topic":"p-recon-trivial-fix","hashtags":[],"change_id":"I30ead8b72cb616d1311ffc81d9cfecb1afc9a05e","subject":"Trivial fixes in recon middleware","status":"MERGED","created":"2021-02-24 11:03:02.000000000","updated":"2021-02-25 19:02:42.000000000","submitted":"2021-02-25 18:56:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777358-p-recon-trivial-fix","meta_rev_id":"5d913d4faf66c0c7e10868a835cb3ecb32f35c69","_number":777358,"virtual_id_number":777358,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 18:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2021-02-24 14:32:22.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2021-02-24 14:32:22.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 13:41:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 14:32:22.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"25f10a98705a26c3c740755b058c8975f5f9ce09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-24 11:03:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c57492ffda29e1fa052276c72dabee4248a1f0ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 13:41:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9130ad1040f34f98975d4a5b3698bd07 : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/de4c8398446a451b8d51f314771ee161 : SUCCESS in 1h 01m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3067e7b85bf49faba82ef2697676fa7 : SUCCESS in 41m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c601101f2e345968c11c836b95919c1 : SUCCESS in 10m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/66759cc5a6d54389aabe85251d693728 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff73754ddfd949b9b7e156077d075fa7 : SUCCESS in 13m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26c82c18f8f74ed68f618eac2e076251 : SUCCESS in 16m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fd7ece2dd4f1449aa71272098db3f3fc : SUCCESS in 14m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d8e36ffde2b64a2bb21d1416c0108e17 : SUCCESS in 13m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5816083c1654d329aafd45b6adccca4 : SUCCESS in 19m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b17b770734a44ac5af8e58d5b990870e : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/46c0e774c1bb4aab8aed59607be5336d : SUCCESS in 16m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c2381d120559475d85c2559232d5e0f9 : SUCCESS in 18m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/40039e3a3766454ba0bab3ebfe63d437 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b3a3dcf497164e20bad26115b39d999d : SUCCESS in 19m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a58626bbad2749e3b5c5dd3dd823ecd6 : SUCCESS in 17m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e37a814baa25427da18c7c523629aa32 : SUCCESS in 19m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cfda96e0e0594817bb08037d27996ef3 : SUCCESS in 45m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/57e87663467f4995bcc5bd141d04379f : SUCCESS in 49m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90471cc34cd24ad8b5587e3b201c21f2 : SUCCESS in 43m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f9dbba33a044e8b93764705232024f3 : SUCCESS in 45m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b67e7c9c67864be09a6c7774ec753707 : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95df372a0b8f4f44bee1dd3f3cca9280 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f0987cbf4f70425784c01a5dfb5801be : FAILURE in 14m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7aee5046420d405891f97886a7e12107 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":1},{"id":"dcbcf981bd368caa76474c39d848326dbcce7342","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-02-24 14:32:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nIndeed it looks unusued.","accounts_in_message":[],"_revision_number":1},{"id":"5b7e52d18f6f817b785912ea6d4b81313cf515c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 14:32:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59740ec356d1ca51a9cd6ae17334613a1f3412dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 22:44:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f73e2945dd409f90606793546c2753 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/8b4b7e654032434d88616834e7c3c3f0 : SUCCESS in 58m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f919f4c20c884333bb34b26c6e54804f : SUCCESS in 50m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/de5eb71d7e61471990266398eef5bbf8 : SUCCESS in 10m 12s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4f34f8fc8d0943e982e51464cf02c567 : SUCCESS in 8m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/343e2b8b1b3646849d26515757780056 : SUCCESS in 15m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df12894ae255400e9c5b1815951a5d0e : SUCCESS in 15m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fca5cd2a5f8642cea34811464aa42dbf : SUCCESS in 15m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/132f730b720747a3a34f32327aeaed1c : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b77f24d5a9d4e02aa79af7e1966ef8f : SUCCESS in 14m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d6daae009a5c40159370ea57267db920 : SUCCESS in 24m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/852c864a72d141ab850ea7ae493c80ec : FAILURE in 19m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b775d661ad7403ba14a6d041ef892a7 : SUCCESS in 18m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5858a44168724a5290764fb79f73bee0 : SUCCESS in 44m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7dfce991c159428f9c11044f4bc8094c : SUCCESS in 21m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8d1a18a2140c4340a7dfb3815f0f1c26 : SUCCESS in 20m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bae10114e3dc463f9ae936272674b8a2 : SUCCESS in 1h 33m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2c6736c042ba4c1999440d4d07a5f04d : SUCCESS in 51m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af6881be5f204a9bae4a868311ea94da : SUCCESS in 40m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b29479262a69422d95593b8c13f97f96 : SUCCESS in 39m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/291ece7c3ff04e5abf74bb5e4e310f77 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32c04b392be407abf54957b7b9cac85 : SUCCESS in 7m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b94350ae88694ad0918e10649c7d8229 : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cd3902e25990b8280d0a5bc9a2a1a7e29b131d89","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-25 11:10:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"172f7b327371a74d520d4d9734ba982e7030e256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 17:28:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5108639820c745469433e22c7efa6e66 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/58880d08118443d7b3f37cf26f1baed3 : SUCCESS in 1h 15m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8e629e23db345518ce3a10e80a1d91f : SUCCESS in 51m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9187e18b4a7d4053915d2f506a62d416 : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69923bcb5fc141f889cd583941a12011 : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d908d097ee4946ec8325be9fe53ec8c5 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2976994ef4f3498abfbe12d3357c1d21 : SUCCESS in 21m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/11f160723ac94b978c08ca8e5ea10d85 : SUCCESS in 14m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/52ff95773f7243679f96e1b065d6ed01 : SUCCESS in 16m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af66545664884b2793d7f91237545172 : SUCCESS in 14m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0205979a45a34668806dcc66ce6be0fd : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea774a16c89f434b9175a4feb2c8e240 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed22e803d109410cb99d74c17044e24d : SUCCESS in 20m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b30574bdb0e6470eaded4623e2772b80 : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cd98dfa431d94ee9b7ae6cc0b6b52eef : SUCCESS in 21m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0f6fc6acb9384caab2bbf791a6b3d724 : SUCCESS in 17m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eea531749a4c44e7ac6bcc1ccf8ddd1b : SUCCESS in 21m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e699cb5751df46feb63c24aace73a851 : SUCCESS in 51m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/167e96497662455b8e1a18a597c03f54 : SUCCESS in 53m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1407fe493814450f9970cf8037d49df2 : SUCCESS in 44m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c6b59769043459585494260c1161331 : SUCCESS in 39m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07e02e333db6485b85a3e998e2fb7db8 : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a64896b13c5040a18b07061d64c257f2 : SUCCESS in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3b4ce40d594456d9d6d3ba51764c481 : FAILURE in 10m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42708eeff3274c2faa0d1183e96eec23 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0e9381c724f8a4b16a8ef339c07bbc00499b2575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 17:28:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ca6e2fedd8e53d474c7455341f5e58c85ec48e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 18:56:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa2ee8f38294ff38b70cf38e645b5b0 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/bf542f0e987840bf8df846dd0f8c870c : SUCCESS in 59m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28449d0ee14741f9b35507e4745e2d92 : SUCCESS in 44m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6befc0cecf5d4042a1a430baa22e7863 : SUCCESS in 9m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/68845bb197f54f668d6e2a221b5078c0 : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/108147ae20c3467db157928df7a61605 : SUCCESS in 16m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e137f251af3048288e2826fb8edfb8f9 : SUCCESS in 15m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f683dbdb5884fcabf259a9579cc5eac : SUCCESS in 14m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a9ed3774b83748068e4d26e0af33301c : SUCCESS in 14m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e68084ab661c4bf689949b8252f3c11e : SUCCESS in 14m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e0a75aa12214d9e9440b404a7c0797a : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ac5d068f4a2456fb850c9a86a0f77e9 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cbf09bc56d0d43d9b78eab7bbc85f24b : SUCCESS in 26m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c029bf1199b74073a6452774ea1e8068 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ef65106ed67a4244946e7a230f4ae754 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ca1c936bb88341328aabc84a009fa4ab : SUCCESS in 19m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/801bba0b2b9e4b8b98009b7490fb1cf3 : SUCCESS in 44m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/476efadd06e64831a13d3de3f9a69279 : SUCCESS in 51m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06ef2d5d19a64bd891dd84e08e6acbf2 : SUCCESS in 40m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62d0c4ee001e4f71bc43497e55acf9ba : SUCCESS in 40m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4bde85b460b4d0baef0bd66e3bb33ac : SUCCESS in 14m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9da9996ba1d4983bb429f8c3c679125 : SUCCESS in 5m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/473e06a468204f968ed91e6561ac6737 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"de67b8484934e89405b5fac1178fb0d40ad5b3b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 18:56:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5d913d4faf66c0c7e10868a835cb3ecb32f35c69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 19:02:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1018b9d12fde42f2bc8c52a77b5a083f : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d4780efb410a4f70b4a8330743071058 : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d8f5db22a4494efdb9c9b40ef972c357 : SUCCESS in 1m 07s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9eac76258aeeaa3287d3e8c467b6ebbfa5e41c2f","revisions":{"9eac76258aeeaa3287d3e8c467b6ebbfa5e41c2f":{"kind":"REWORK","_number":1,"created":"2021-02-24 11:03:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/58/777358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/777358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/777358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/777358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/777358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/777358/1"}}},"commit":{"parents":[{"commit":"a3d77cac071ec3ba890bd13f065824cd43179000","subject":"Simplify interface to sharder shrinking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3d77cac071ec3ba890bd13f065824cd43179000"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-24 10:59:38.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-24 11:02:47.000000000","tz":0},"subject":"Trivial fixes in recon middleware","message":"Trivial fixes in recon middleware\n\nFix docstring and remove unused mount_check var.\n\nChange-Id: I30ead8b72cb616d1311ffc81d9cfecb1afc9a05e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eac76258aeeaa3287d3e8c467b6ebbfa5e41c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eac76258aeeaa3287d3e8c467b6ebbfa5e41c2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
