)]}'
{"id":"openstack%2Fswift~777369","triplet_id":"openstack%2Fswift~master~I9f9313df2a7fc02e843053706ad06aebca0aeeaf","project":"openstack/swift","branch":"master","topic":"update_stats_sharded_split","hashtags":[],"change_id":"I9f9313df2a7fc02e843053706ad06aebca0aeeaf","subject":"Fixups for direct_get_recon tests","status":"ABANDONED","created":"2021-02-24 11:48:05.000000000","updated":"2021-02-25 05:37:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa61bdbc3cf3ec7ec60063e1754184bad7824f40","_number":777369,"virtual_id_number":777369,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-24 13:22:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 13:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9df1f65d9526f722f9af82d658b882e2c22ff3de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-24 11:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c44e375a18a954c644d1dda466c3c089e6aa840c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 13:22:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bb657e21eea4d2986f2cbf9c7edfb09 : SUCCESS in 9m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea5fe0121f114ff59a15d176e2f446f4 : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/349a5e7704194b12af32b41e3e157243 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bafd254404e0452db414bea200c56ed7 : SUCCESS in 15m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/265c9cc0bfde4eb49789dc0397a7e82a : SUCCESS in 14m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f8e89baa384641c788a1a6a0e16691c5 : SUCCESS in 17m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/196aeb712efe40d28543fd14f9a7cc53 : SUCCESS in 19m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a9c0f6d906b44fcb4e5db992620a6c4 : SUCCESS in 15m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4866c022cac480ca9b33cea15883e9c : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f05b00fa7f44ec98b8a864d3fd56f37 : SUCCESS in 15m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc0577632ca44280acb702bdf5281bfa : SUCCESS in 19m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cc76ce976a74461b80691cca8bd4d445 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eb6c1ef0ee864643bcf0093542c50bb5 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/75c78b7181b44b99b8f89ab203078cd3 : SUCCESS in 17m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f013a7e725b44cd88537eb95270d1298 : SUCCESS in 38m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba69b26226144b3fab8c2cf6c497a960 : SUCCESS in 38m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d32bc6b84eb46d2b76df3d28ab8cc18 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3cf77df531d4de785d5f289e6d5e9e0 : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9022435df4a84dc1b2369a87dd07ff2f : FAILURE in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa61bdbc3cf3ec7ec60063e1754184bad7824f40","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-25 05:37:48.000000000","message":"Abandoned\n\nsqashed into parent change","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e6c826d747ebdbc3d25c95e9705e58d699d8b88","revisions":{"7e6c826d747ebdbc3d25c95e9705e58d699d8b88":{"kind":"REWORK","_number":1,"created":"2021-02-24 11:48:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/69/777369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/777369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/777369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/777369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/777369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/777369/1"}}},"commit":{"parents":[{"commit":"01643a339dae1c50b68a0e15f3c26405ddf529ff","subject":"Plumb sharding stats though recon middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01643a339dae1c50b68a0e15f3c26405ddf529ff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-24 11:47:28.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-24 11:47:28.000000000","tz":0},"subject":"Fixups for direct_get_recon tests","message":"Fixups for direct_get_recon tests\n\nChange-Id: I9f9313df2a7fc02e843053706ad06aebca0aeeaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e6c826d747ebdbc3d25c95e9705e58d699d8b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e6c826d747ebdbc3d25c95e9705e58d699d8b88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
