)]}'
{"id":"openstack%2Fswift~777405","triplet_id":"openstack%2Fswift~master~I9de032d309befea2ff8b752797947b9fc3ed8698","project":"openstack/swift","branch":"master","topic":"cors","hashtags":[],"change_id":"I9de032d309befea2ff8b752797947b9fc3ed8698","subject":"DNM: verify zuul cors test fails","status":"ABANDONED","created":"2021-02-24 15:31:10.000000000","updated":"2021-02-25 11:08:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cde1816a3bc6fe8bbbad4700c7752c32fd8bf4d5","_number":777405,"virtual_id_number":777405,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-25 00:00:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 00:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6ebd02605ad2bbae577045596cefb921e9d66cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-24 15:31:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13f1a4d243710fdcdbdd829989f4341ba939d1f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 00:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e22e37a3a8e43cc8b09678b703a4a6d : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/978b15823bf0416a82c686ace62d9dd0 : SUCCESS in 1h 11m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/011f6320b14b4d4fa13c7cacecdf9e1a : SUCCESS in 44m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7cb067fa79ce44d29fc3c54cb4c15a78 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6fd9a27bb30e40d09a0b4919b6145438 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2390fa54ac0f438ebf067c01b65d65e5 : FAILURE in 16m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c67426557ab641e88496e53fa3e24c3c : FAILURE in 17m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bfe5efbdae1247fcbe54756c0cd55657 : FAILURE in 20m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2710871cef8a422588a1a3cb13b056c4 : FAILURE in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d086051db634f25bd270698fc4f1f25 : FAILURE in 17m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9fcba4364b5b4d1ca4a2d5ddad7cce33 : SUCCESS in 21m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a6a8d7ae77b34489ba0204131dd4aee1 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/becc0c0d4ec84d818850ef706324b8e7 : SUCCESS in 17m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ed4ddecde1dd497b874c77a24c800d2b : SUCCESS in 19m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/435dc129a5f749f6a8c542668ad1488f : SUCCESS in 20m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd7ade62049d4cc884ddf6d6852c0367 : SUCCESS in 26m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7cfac1e3d8e14b2486e971b3b8d5b204 : FAILURE in 13m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce656639834d4494a0bf3147f915c121 : SUCCESS in 30m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1caa03d7c1974d87b20820c2754e331b : SUCCESS in 53m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3d83969aa2b342ab934e14306cfeb9e1 : SUCCESS in 1h 01m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04479b61f513460a93bd52d6fd681a37 : SUCCESS in 40m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/231db3bac6b34dfdb87ff047b327afc0 : SUCCESS in 46m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6003dd0ad964462b935fa503e85c00f6 : FAILURE in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f473dd0472be4b4b866a7a2d7574a2fb : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6d6048ac3d241aa85339b6bd94a7da5 : FAILURE in 11m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8a9b349fd6e4d48ba67425d8799ae03 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1866ed561d168165062ba80dbb6e1ded79ab2aa8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-25 11:07:58.000000000","message":"Patch Set 1:\n\nswift-func-cors failed with:\n\n  xvfb-run python /home/zuul/src/opendev.org/openstack/swift/test/cors/main.py --output /home/zuul/cors-test-results.txt all\nnon-zero return code\nTested firefox in 2.9s","accounts_in_message":[],"_revision_number":1},{"id":"cde1816a3bc6fe8bbbad4700c7752c32fd8bf4d5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-25 11:08:35.000000000","message":"Abandoned\n\nno longer required","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2dcba3c1cc339cb5a94552cda9d6eaf6e4e9b283","revisions":{"2dcba3c1cc339cb5a94552cda9d6eaf6e4e9b283":{"kind":"REWORK","_number":1,"created":"2021-02-24 15:31:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/777405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/777405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/777405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/777405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/777405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/777405/1"}}},"commit":{"parents":[{"commit":"1b6795a905857264f885c2ca1287222f3b3f4869","subject":"Add some functional CORS tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b6795a905857264f885c2ca1287222f3b3f4869"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-24 15:28:35.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-24 15:30:53.000000000","tz":0},"subject":"DNM: verify zuul cors test fails","message":"DNM: verify zuul cors test fails\n\nDeliberately breaks cors to verify that the zuul job fails.\n\nChange-Id: I9de032d309befea2ff8b752797947b9fc3ed8698\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dcba3c1cc339cb5a94552cda9d6eaf6e4e9b283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dcba3c1cc339cb5a94552cda9d6eaf6e4e9b283"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
