)]}'
{"id":"openstack%2Fswift~777604","triplet_id":"openstack%2Fswift~master~Ica983aa13a7bcd07bc73c68f3938b190c4bbaa3d","project":"openstack/swift","branch":"master","topic":"p-sharder-context-done-tests","hashtags":[],"change_id":"Ica983aa13a7bcd07bc73c68f3938b190c4bbaa3d","subject":"Add unit test coverage for CleavingContext.done()","status":"MERGED","created":"2021-02-25 15:00:17.000000000","updated":"2021-02-26 07:11:36.000000000","submitted":"2021-02-26 07:10:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"777604-p-sharder-context-done-tests","meta_rev_id":"26d77fe16b6136ab5469fd3e08478218cfd774e2","_number":777604,"virtual_id_number":777604,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 07:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-02-25 17:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-02-25 17:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 17:21:08.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-02-25 20:16:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-25 21:44:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"c66587efd26768253c8f88469d97828f4cd70c6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-25 15:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d8eb3de68016ef382aa8738170eea154fc124b6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-02-25 17:21:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eeb8c367087bc70d1db006982983839f93d2949d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 20:16:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69bd664e7f84454aecb36836451e923 : SUCCESS in 10m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27c386fe6cf34735839ea89e4328e139 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/07737a81b088469485d23ccf80426735 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4265b90a36994592adbee93621e13ebb : SUCCESS in 13m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/adf6d2bf327a48419d682509dc37408f : SUCCESS in 15m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c51789820be54e07abcf8f4f8d889cd4 : SUCCESS in 15m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0e5efdac154d42418cb15135ee20dbdf : SUCCESS in 15m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7306f936a77641909396b955cdac1b8a : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b7b4ac65fcef497bbff750874b650fee : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc9fab72170d48778a4e0c801e5b5d63 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa28dbeb368145aeb23fa64bc354a1cb : SUCCESS in 18m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f5178958fe2c4ca780ffb5eb0f4a6b1a : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9513c7286c554cad9f5d3e5dcecdac94 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/040b13dd5d1f4fb78957d36a9cd25b56 : SUCCESS in 21m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/079b9225e8194a4b8b2d7f541cec7342 : SUCCESS in 37m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/29b099e142384a98861ab23945b75452 : SUCCESS in 44m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ae5f863eec643bf819427e859e15797 : SUCCESS in 15m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1ed75ed9a647efa9ec29ea4c4bbd13 : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0eae209dd167463ab9bef11b73a87764 : FAILURE in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"179566caf52629c3d8436e57b74d7ae3ccf6cb14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 20:16:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57766715f8fcfa8e9d0c744e234086c75979c932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:41:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0415f06de2423b8ce128ab4e4dddd9 : SUCCESS in 8m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/47c62da7a0de4a14b6f5ed6386ccfecf : SUCCESS in 10m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a62224b751864d4dbcfc77715dd819e0 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ab5390701f848538d27266489b5d83e : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b5e3b0d4e7f4ccc8326849804af1274 : SUCCESS in 16m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/928fda1db876435696cad1e646c2e4f2 : SUCCESS in 16m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c5a1a7bfac2b4f3991a7a449b67c61b9 : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62f3c9b488de4781810e817b61c9d058 : SUCCESS in 17m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/523fb08ee99542989f2625799d01f527 : FAILURE in 20m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20ba72780547448bb9582a7c501ec699 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/41f8153f55a34357ac41a16aeb81777f : SUCCESS in 17m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/300835f4f55444d0bd02307d97b8a34e : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0633122f187840c6b682ce9bb2400c58 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d390e6d28cc840d9b8f71071e1ce9fc5 : SUCCESS in 18m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/18473bdd0a5c408db1ad37c01c2d9d8c : SUCCESS in 44m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4701e3fe1ac14f999de414083a31ebd1 : SUCCESS in 48m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cab54c5a6804ec6b6c055f2a8669a96 : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b56f174713144ffc987f6dd30ed8f4b2 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"98324f0d541b29202b5e0337cdaff4759ec5da0b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-25 21:44:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2246aab1f319fdbcb151b4f436534c3fc0cdad0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:37:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab302a818a504e63a43e003c7961a619 : SUCCESS in 8m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e4baf902963494eb1c9b98cd9122f4d : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/95982c4f476c43049b17ed75a95349d0 : SUCCESS in 7m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8fb7f56d67d448ec97f84295c9706d14 : SUCCESS in 22m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74201f6da69b4418b0725337952ea31a : SUCCESS in 16m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9912fa1723054658936300d13bd9f797 : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/38bfadd5d6a346dda7db569e3f752f17 : SUCCESS in 16m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6dac7c9a387e459dbee4a5806c88b224 : SUCCESS in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8afaeeeab374dd095c48fa1a8ad8578 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/222a18105ec5413997c96d97362f16c2 : SUCCESS in 31m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/32f121f3fa734263a5b5c2ac742e3eb2 : FAILURE in 1h 00m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/43e4f85a2ab64bec96d803409dbf77af : SUCCESS in 32m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/70e5449a4e65461dbcdc5e15c52558e4 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0716b4e8e1a242cd88b737b153af4fbd : SUCCESS in 17m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a34c532c09874b70b8a3cd0f820a4b46 : SUCCESS in 49m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f644866d57447db9387be534d311557 : SUCCESS in 37m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb71e0bc67ec4ad68505132d5e4705ef : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/161a973f33cb4d84ad9a9587d4f8ef55 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87c1adb403f942cc87f68074f2ed122e : FAILURE in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6733598b2516efd8eec714fbb731ff55142126b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-02-26 05:31:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4cfac4dff31f9111d53cd5429a929171f03ca210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 06:19:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a2e903101b4d23ac7d05fc7725b07c : SUCCESS in 8m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6a6e7474810f4abcbfcb38f84411e1ee : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e7b6f89338846c2b64a24551be702be : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5360e11b1bb841ebbf965abd759b2c00 : SUCCESS in 15m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3b2510e86f4e4606ab8d89f4b61eece4 : SUCCESS in 15m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d9eefa95d6a2405eae4d6ae921a9ad07 : SUCCESS in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/546775a5c60d45f4ae57b0ecaba861f2 : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9962f5e35fb142a794093caa2ae64e6a : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/127147247fd24f03bd4690d527b8edf8 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10e57477064a49e09109bb07d64eda23 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd06a365770541f38bc6dbdb8ba78b55 : SUCCESS in 20m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5705770dd8d44efd8c685fa545cb1b70 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9f57424f35734dec8f554082751d8ea2 : SUCCESS in 19m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e20bbfbdd53f4cbcb4b8306078178b91 : SUCCESS in 17m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b417d3124ff492fad33c509d1d0ad45 : SUCCESS in 42m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3d4b7dc04cc4b948a4a974896515417 : SUCCESS in 39m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e4f57591f6d4cd2985e626b08d77474 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebd72ee961343eebbeb7c9dd01b1ac5 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d07cf89c81d841cc9f107bd3126960d8 : FAILURE in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30a35faecfb78a1b4ff707bb05bb6f338fe45b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 06:20:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30aa096fc3600468daa140b28e8c5304ed4d4e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 07:10:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b81e65e793492e86910027be4e1ecb : SUCCESS in 8m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/be60db9896394770897124ed3f134913 : SUCCESS in 11m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/674bba6a958547fb862cf8efecee7725 : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/505e14bb5f8a4da69d78be169770f000 : SUCCESS in 21m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17e478e84ae748fe90e68cd3bde1408b : SUCCESS in 21m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e5fb1a5d0a814a19b663506ed057ae70 : SUCCESS in 14m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bb715be63e6c41a5ade40fc8dc3d7256 : SUCCESS in 20m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/620ab6389dc64083973d5def5fd78c46 : SUCCESS in 21m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2cad2a10f8945288d3dfe3a636eb770 : SUCCESS in 33m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d885aa199234d2393eac7c97c0724a8 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc5115609ae645038401a28582c7d4fb : SUCCESS in 26m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0f45d481dd6b4fe3bf2a0c09e8223af9 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fc816b13f1f24cfda59d0bc7cc56d9a8 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2a329b33ab504d688c11537cf32a6a58 : SUCCESS in 18m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a7eab58bae9241a3bdcaf31360269c50 : SUCCESS in 41m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/995d1e6c838a43bdbe9c55c6d34fdf39 : SUCCESS in 47m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a9b0d270a8a42108fe65d708a14f1d3 : SUCCESS in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81173d29a8d74882880c2b485d851bbc : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"10d960c1f711df69160253b976f1f2121cf13cc2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 07:10:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26d77fe16b6136ab5469fd3e08478218cfd774e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 07:11:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10664f26447f4dd696119e5a5c83472c : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a2848771d21b494e82a623c99406f65e : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/23dada623af64b939814707036ebd00f : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9f6e760bc2b70e025edb149417862ce77689646","revisions":{"c9f6e760bc2b70e025edb149417862ce77689646":{"kind":"REWORK","_number":1,"created":"2021-02-25 15:00:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/04/777604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/777604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/777604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/777604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/777604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/777604/1"}}},"commit":{"parents":[{"commit":"a3d77cac071ec3ba890bd13f065824cd43179000","subject":"Simplify interface to sharder shrinking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3d77cac071ec3ba890bd13f065824cd43179000"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-25 14:59:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-25 14:59:00.000000000","tz":0},"subject":"Add unit test coverage for CleavingContext.done()","message":"Add unit test coverage for CleavingContext.done()\n\nChange-Id: Ica983aa13a7bcd07bc73c68f3938b190c4bbaa3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9f6e760bc2b70e025edb149417862ce77689646"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9f6e760bc2b70e025edb149417862ce77689646"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
