)]}'
{"id":"openstack%2Fswift~777717","triplet_id":"openstack%2Fswift~master~Ie8e2b03f93ace0283c7a759d2404022208c0c553","project":"openstack/swift","branch":"master","topic":"add_config_to_manage_shard_ranges","hashtags":[],"change_id":"Ie8e2b03f93ace0283c7a759d2404022208c0c553","subject":"Add name to compact output","status":"MERGED","created":"2021-02-26 06:20:09.000000000","updated":"2021-03-13 09:22:21.000000000","submitted":"2021-03-13 09:20:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"777717-add_config_to_manage_shard_ranges","meta_rev_id":"1daf529a5fdf899fcec7ec6b831881e53c5fab2a","_number":777717,"virtual_id_number":777717,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-13 09:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-03-13 09:22:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-13 09:22:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-12 19:18:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 08:12:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 09:48:41.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-03-12 19:06:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f87b264be795059c4c31dd2034b798240b00c428","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-26 06:20:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690a0438b406f22aedc5a71c78ea3b1fab355a4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-26 06:22:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fd8912d46634d5f152ab0fab32c3f8b28bc8c40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-02-26 06:25:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8097e29b8dddfeadf85fe01e436d4709cf9471b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 08:12:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16fadfa1e7b4d8b9e23930874a39abb : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/0494a22dff2442b894e69398532afb42 : SUCCESS in 1h 09m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a65346877144e45a4962135251b5855 : SUCCESS in 35m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c722c0a5a1e46cca2862037734e33a5 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0eb6fb89a9824587b51a98c2fc0efff7 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eede5146bea8461cbc9b7aaa6f222e86 : SUCCESS in 14m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d718f115e824051bed58cc615719d0b : SUCCESS in 22m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a00933212dc4cfeaf1780c8bc21dcfa : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5316e40cd3da41febd1e3ad50d6f8d78 : SUCCESS in 14m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/09dc91b42c5947f3969f86d9f2a8c605 : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d3d757814b074c4daeb6034fe2325510 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c536fbde515409492cb55f18ca445f9 : SUCCESS in 20m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84b00b1dc893432ab112fb659589b211 : SUCCESS in 20m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/61df41e397624454bcb69d0354608a20 : FAILURE in 19m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8be94ac70ff34611ada02c9f0fa0e9f6 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/af6f7715978b4b1eab697419cab77f55 : SUCCESS in 20m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3f534bc4eff4cfa8d5ba4da77f9eaa3 : SUCCESS in 23m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85ef1d2e91064c7b9736e89595d73258 : SUCCESS in 49m 10s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d3134be7e7ac42b08ea1302ff30b2823 : SUCCESS in 48m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab8e6d17ce10487a94bdc6e4a259d649 : SUCCESS in 31m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/542255feb78b4c54b2ae32d630fca383 : SUCCESS in 43m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408b596289d94e57bbd3e147bdfa44f0 : SUCCESS in 22m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08efcc70542b4a8daf4b25e2da9db34b : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77392a978e68464ea6e356a928b3fe7c : FAILURE in 10m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/112b20cdcbf047d9bb805b33372f344e : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":3},{"id":"1f06b24b7b1ed280b964fb42cd3ab566198dfa7c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-26 09:48:41.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nYes, name is a useful addition, and maybe we should always print it?\n\nOne comment about the format inline, mainly because originally I was trying to help my eyes when scanning the output :)","accounts_in_message":[],"_revision_number":3},{"id":"2ead76173780f5c81d041659acb58a95b835059b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-01 02:50:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"28caeb4adc166d4f2511bb06c5726388c96e2362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 04:00:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da46cbd9adcf4a409228c005eea5ff95 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/14c5771bfe604f5e91666c87ef4745ef : SUCCESS in 1h 04m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20cd7a6d5d724110a9071e90ebc26217 : SUCCESS in 47m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34deb7bfa04847f788d9e647ff930b95 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e39f5c417d5f46ba8ea625b4b770bd44 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/72f16c345cdd4c588638c2ec30b3eabb : FAILURE in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ca848c02fe24e09b6a1983c13ce5679 : FAILURE in 14m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ceae1db3d9f2435aaec40ae6cfa77920 : FAILURE in 14m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0e5fd91a5e9847ccbe42ed28e1d0fdc5 : FAILURE in 14m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35299fca4e74484b9cac9cc14f74ee33 : FAILURE in 15m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/98b6b4275d7443cb9c4cc9d61eff1fec : SUCCESS in 25m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe1cc4e621364decb8db786e99ddc199 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/feadedc455c848839f7b925b1cd7d03f : SUCCESS in 17m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9b0ba33ff64b49ef9264ff2adaeda4fe : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eeed8d023acb486c8323d6272ffb5357 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2874239051fd48b8a3a67d59dcd16a9f : SUCCESS in 18m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/830b45006d7646b98f338d02a7801cb2 : FAILURE in 31m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/228e3c19acc94295b65ee86017506ce1 : SUCCESS in 50m 49s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/96a01482bc554bb9a4b14b0fe0562f32 : SUCCESS in 45m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f124c02f3e44a649fa46c732d411890 : SUCCESS in 45m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dedbd729bd794a419bb0411954293af4 : SUCCESS in 47m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eff9de12358c46d9a10d426e7c0482c1 : FAILURE in 19m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3014598bad7a4968b769c8f1f7070f96 : SUCCESS in 7m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a643863df4f44ec8cf1efd63e44a0a0 : FAILURE in 11m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49203c0b98554d5980a45aef4ce2547d : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":4},{"id":"79489806c45a72e89bddada093676ed6fa9a90f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-01 09:34:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ffc2f94b23b4a3b5d7c0d3310576fe0b122de70b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-01 09:35:34.000000000","message":"Patch Set 4: Code-Review+1\n\nI guess there\u0027s a unit test that checks the output","accounts_in_message":[],"_revision_number":4},{"id":"4ecb219c88ce04f99178c34b64a8cbdb378d24b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:44:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df95f3353df24bfca8963348f3fb8116 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/148a1b817d694881b37be6ca2fcc2c32 : SUCCESS in 1h 13m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3a25f5219fe4c178a972704b7c50592 : SUCCESS in 49m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e20e3be4d754e1aa9bcbf45424736f5 : SUCCESS in 9m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b08fcae0f4c428daea765c5d553888c : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00ef3ee5fe2b4e14bf3fc3b8a97cf967 : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ce437ee8b214f7c9bfdeef4c223b73e : SUCCESS in 14m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/650f68d3beb741098ac0ec391509c37d : SUCCESS in 15m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/758486e699bc4e96952d46bb0cde02cb : SUCCESS in 14m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6790099068a749bba6442d0239187d71 : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/435cab47bc1b4210a32f2db9f165841f : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8f1c6dbd6874a018ff2db3dc107de89 : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fbbc03984da74b77a97ba6d5ac026923 : SUCCESS in 19m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4b379cba61da428eaf4ced1f690d0e4b : SUCCESS in 20m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7240108d64b6407cb2ff6b56d34ec07e : SUCCESS in 19m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bf7a464e5b664a56a8af11a5722b473a : SUCCESS in 20m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cfa828c4861046f5834e3139189d050c : SUCCESS in 21m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d67a223763a14a838138d12e42491778 : SUCCESS in 46m 54s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/495e714bed96478fb84b707b285eea7d : SUCCESS in 45m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e33819b47c59402a91247e00a3c3bede : SUCCESS in 39m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4225537f98074b5e90764ffa5c49b903 : SUCCESS in 38m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44189cd6ce6144c49e1b867c82b365f7 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd784b13e03431a92af0772d2c72dd4 : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7983cf9bfdfd452486db83b59c9baaa8 : FAILURE in 11m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0b99b12d0e9430bab96c976566e95ab : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":5},{"id":"d2123378d381a27ec67b7d55eed6e0cf46c169c7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-01 12:40:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fde0318e7e4b8c5233803d3361ea2978dbd00e2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-05 17:19:50.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"2cdbd5bca5b23de11c4a59ddcf49db4c0281e0f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 00:42:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58319dafef74908988377477d541aed : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/4f9c039c864349889fd014ed1635dce9 : SUCCESS in 1h 00m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/36f33e42aa594e9f89a4eda49a884ed0 : SUCCESS in 46m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0edd8b37ca004ea5bf8f4e425a95c33a : SUCCESS in 11m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/04e8b6529eec4a07b85120d9080f84e0 : SUCCESS in 9m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6564984dfaa5464caa5f55d654cc2f9b : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7d9696fff5f41b19ab72e45173575f2 : SUCCESS in 17m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4b8b95396444d8b842dcc30df47fcd9 : SUCCESS in 16m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4b759caaab4341a48e361512e7caa567 : SUCCESS in 15m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/db72aa92c82e441793274b22bd6d3f24 : SUCCESS in 21m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39988fa1d58341689abb49bc818c7a65 : SUCCESS in 22m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07ee4294a5f545f7be04c2251b797062 : SUCCESS in 21m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/252940e6e1dd45f5aa1eeefa8b5787f0 : SUCCESS in 18m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/508f37dce0f849b19b8f0b7edd59ff72 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e8b6734f870843bdb48fe334b1b11b1d : SUCCESS in 21m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d5e15194d8ac4e7695bb48c9455e77a0 : SUCCESS in 25m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eceef4503a9643a5b16b300434a09431 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/23c5653ac37544a791eb2d3c37ec9845 : FAILURE in 22m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3baa600dede042d39ac6e9478fed6769 : SUCCESS in 52m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f1749212b02d4197ad53af7f8658c4ab : SUCCESS in 54m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15658a750dd749e9aa9ae3966cb24923 : SUCCESS in 45m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f79a5a0ee0674f41a9ff827e748c8836 : SUCCESS in 45m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5d0cdcfa19c4cd980d7c9f88f234861 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edabd8c167bf47e48ad64b90ebb5ee49 : SUCCESS in 8m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11040e2206fe4cdbb1c4ca181604e1a3 : FAILURE in 10m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f53f39f7781f42d8b775a7d299b6b976 : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":6},{"id":"f91b8fc761885ada32b1af11a2961b2ed4fa1174","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 19:06:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e68ec194b403e81cd96495fe43bf1f736978c5ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 19:16:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b09bd87a68257444d2cf00fb63916d7aaf5f33f6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 19:18:35.000000000","message":"Patch Set 7: Workflow+1\n\nRealized this needed a rebase. I think I maybe like the alignment more in https://review.opendev.org/c/openstack/swift/+/780313/1/test/unit/cli/test_manage_shard_ranges.py but it\u0027s not worth blocking the improvement.","accounts_in_message":[],"_revision_number":7},{"id":"c84942f0b9a1410cdab05ba15f2701a4774facdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:34:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c1c9ab41a74cef9004ea006f93f46e : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/d22a9c5c5b3e482a8f5b534f291fb04e : SUCCESS in 1h 09m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad7cd2141572409eba890ed0e93f9e88 : SUCCESS in 49m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b877e803c3445449a592f156af83ea3 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba2e38f44a8c4b87ba64db7720432c61 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/687b5743a95446f2a01240020907bc87 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6408b226b11040899176210d5573b7e4 : SUCCESS in 14m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/33bc69c7bf654cafb083b3616b1bfff4 : SUCCESS in 14m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4c0c1dfa2e554c2caf6826ee4aeb8ae2 : SUCCESS in 14m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2fb8b7e080ae4c1b92c6d59a0642da7c : SUCCESS in 15m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17c2bc801e8a4cb08cefe965cfebcf23 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51c3425e46ae484fbfc42694de741ade : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a1e6947d318d47918e201d2c7e363a1a : SUCCESS in 16m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ab9d8fc7a17e4cf1bf5911a76cd39578 : SUCCESS in 21m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/53c18caefe674ac7af49910ebfa381a5 : SUCCESS in 39m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/838770be912540af8491b2e12c54da37 : SUCCESS in 22m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad3ad9ec4913411ba46b723d45b99f9d : SUCCESS in 7m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4f419911fa245879d7e5540f3cdc756 : FAILURE in 24m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a66980868ae34c6ebf5e7bcfd700e02b : SUCCESS in 53m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b1f6be04c7c94bb0ab68a0e949283c29 : SUCCESS in 44m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dedadf4043394ee0b96f0ec950dff3e7 : SUCCESS in 38m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd3aba91b55a4dee8c4babedbf382655 : SUCCESS in 37m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e18d25705a64433a15b1b703782d9cd : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/106ba165628d464fa66304f7c0bc4779 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/889bd26fb53a4724aa31778d511d76cf : FAILURE in 13m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd6d8c23c5c44c348074f5e27f17fd6b : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":7},{"id":"d60e8de0821acc0ee9d8232897e8dfb1db71f69b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:39:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a110ff6532e8e36256a940ba18eabaab647be9b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 21:16:16.000000000","message":"Patch Set 7:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9f5e7fff65371c2b63e50b1e6c7942c567dce1e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 01:31:14.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a50b0277a6b429e825dc7f4ee99d0bb : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/1fe3a6f2141e4836b6e01ffcb2b13260 : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8ae3484aec14c0b966a4fc7545a09e0 : SUCCESS in 42m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7d9af5accbae457690fa763010a1535b : SUCCESS in 9m 43s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6c0d6ed0ede24fcd9ecc7b4d5be767cf : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19d14e73f7da43ebb60c5d76299d350f : SUCCESS in 21m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad8332d38db24ae6b068d79619b42a92 : SUCCESS in 22m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2debbf656188481cbaff87c4b199f397 : SUCCESS in 16m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bad69b2a4acb499bae04996ae32ea0a3 : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/817964b85a234e379f82cc6acfa864ca : SUCCESS in 22m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2d9f4795aba42e783215b1a0039ccc9 : FAILURE in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c863d723bae4b00bc5307e4c8db27c9 : SUCCESS in 22m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f34f4db9c82d4488a4308492ab719065 : SUCCESS in 20m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3836d47b799c47d391aa87d8b895277f : SUCCESS in 20m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ed25d6a46a2141ec984bc383d9e5cd87 : FAILURE in 20m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1bb75a7873b04aa296f15d66c9a64e48 : SUCCESS in 27m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/22a2f98942d44018b8f7ed551ae95644 : SUCCESS in 6m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/950401b686b64f168c14a45070bdb3cc : SUCCESS in 50m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9609896ad65542119abebc35c4de4b84 : SUCCESS in 51m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/763090a27d9843ef9e8439c9fb042407 : SUCCESS in 42m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6756b738f4084a01965afd02d947a700 : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a314907851cb410fb807af747b27908a : SUCCESS in 26m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e277ae29c745bc8a7c21711864b271 : SUCCESS in 7m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d0aa722568442a0b0ecbdf4191bf38e : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":7},{"id":"c601d2ff985f40b5d482c52b5874de913cad3644","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-13 05:23:50.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"20de7192d5ccc8a10dfd0639270848d6e3b731ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 06:44:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be29f4b5df14c82afd11a4617815f0b : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/9bb5b87ea4d54f7aa941d7a19b14007f : SUCCESS in 1h 11m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/441cf26cda4f4f82ae8c986dc94c220d : SUCCESS in 35m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0961bc8f362d4070adc4318bb5bf672e : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/80663689b3d44a02a20eb8a6913812b5 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78e5e1a69ec448e9aba4a644be014a7c : SUCCESS in 15m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4273d1704de244b0b53eb18c503f31fa : SUCCESS in 14m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d97de35bc1be4fe99234fc7cf2c16858 : SUCCESS in 11m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb2c2881e09e43aa8d57db49e42bcbe5 : SUCCESS in 14m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82631a7b3af54d13a2de546a3d7833f2 : SUCCESS in 14m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36c4e5ecf62043c0bd07c8a048b5b45f : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43ee6226a77940b1b79bf37b3766b036 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/861a4d13529b40eba89364b0f0be676a : SUCCESS in 19m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8aef228c1bc14ab5821998f7c20daeb3 : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/94c88de1aec1441b80213ede2b7d186f : SUCCESS in 13m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/448f147b3577413c8950c3319469429d : SUCCESS in 19m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8125e52938e4cffbd683d21cb9f98c4 : SUCCESS in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0945237f38e641bd94af5eb112c123a1 : SUCCESS in 15m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88ec3d5f3cec4e97bb2ff081f969b33a : SUCCESS in 49m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5afe402e7ca843e79fac8efd5b2f1e46 : SUCCESS in 45m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c6c9868704c4205b48f2b2557e01f06 : SUCCESS in 44m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33fe941f55a543399499b58443d8a930 : SUCCESS in 43m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66230c22bcf34200859341af8aca833a : SUCCESS in 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18e0aa5120644f88a6f1776328d7023 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f493af3a1515433ab82944bc7a182c12 : FAILURE in 9m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a28ccc6a23194773bb8dee2fea2f5b92 : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":7},{"id":"89641dcf07b604af53ea3d1a5fa540b57db0b497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 06:44:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"accc1c9f46b6685bfe241b20656a9af720f64474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 09:20:07.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255bd3ede57d4177894f43b1f62b03ed : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/f5a706a2014f4c7185fad9ff1dccacbf : SUCCESS in 1h 03m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65aedd50d39d4cef9d6eacb39e72d130 : SUCCESS in 46m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/33f12762243041a2ba2f5aabb66253e6 : SUCCESS in 9m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6fcedaf1595745d7b181bde1c5063202 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f55decec3b19410c9789cbfb874a9c78 : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6d4fac3406e5453b95aeb387c77e6633 : SUCCESS in 15m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/421c0b8b071a45069adcc79a1d395f62 : SUCCESS in 17m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50bb70b190294cc68212068a3278c3de : SUCCESS in 18m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a940ded191be4ee09b6a7db58bced752 : SUCCESS in 15m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19ea165d6f624016ae02d0a02fa53938 : SUCCESS in 20m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dfe6a84e5ce14f1a8787ef46a313f3e1 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/35f1722346ea4b6fb8ac1571a628ab18 : SUCCESS in 16m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a378bfa3d81943a1b2e94a76be7d133d : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/956eb798f0574a8b8eb557683784b3a0 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6b76845123d4485c94bb0775be1d5a7b : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/602e818a8ed74dbb969f9ce0629f585d : SUCCESS in 6m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f40f63c881fb4214a14061449d20f332 : SUCCESS in 37m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f320e998836e422fafcdacd2cb0992f8 : SUCCESS in 43m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e577842047f421791544a33c482921f : SUCCESS in 38m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48157d108c8b47c496b03003a627b880 : SUCCESS in 39m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcfe989059fb462e885815965ca4a959 : SUCCESS in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82afbda9fca2405da1fbaf2578e03fa3 : SUCCESS in 5m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffd6ddc0d86a4366920c164cc12c3657 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":7},{"id":"9b2c3082f510ceef376c7153f45dfc671d5e7c05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 09:20:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"5447ef3f2286a1e9d899dac62cf187ad0b053dd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 09:22:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e76ba43424549dfb2d1e3a7fb2c5748 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4d47fae13f534ca6b0fda59d050a036f : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/482a3447e4d0483eb688ac474e4b59e8 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a045f90b4d7ccc2da4cd52883c04c176b9ec920f","revisions":{"11a476bbc4641ea1d6972ee15bd0813727028807":{"kind":"REWORK","_number":1,"created":"2021-02-26 06:20:09.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/17/777717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/777717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/777717/1"}}},"commit":{"parents":[{"commit":"b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7","subject":"Add a config file option to swift-manage-shard-ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"subject":"Add name to verbose compact output","message":"Add name to verbose compact output\n\nWe have a verbose option in s-m-s-r and when you get to the final\ncollapse into root with the compact command, it can look a little\nscrewy:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the\n  sharder runs.\n\nSo if you pass the --verbose flag the name is added to the output:\n\n  $ swift-manage-shard-ranges --verbose  --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027AUTH_admin/shardme\u0027\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nThis also works any time you use the compact command:\n\n  $ swift-manage-shard-ranges --verbose  --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-0\u0027\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027e9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-1\u0027\n    objects: 0         lower: \u0027e9\u0027\n      state: active    upper: \u0027j9\u0027\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-2\u0027\n    objects: 0         lower: \u0027j9\u0027\n      state: active    upper: \u0027o9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-3\u0027\n    objects: 0         lower: \u0027o9\u0027\n      state: active    upper: \u0027t9\u0027\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-4\u0027\n    objects: 0         lower: \u0027t9\u0027\n      state: active    upper: \u0027y9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n    objects: 0         lower: \u0027y9\u0027\n      state: active    upper: \u0027\u0027\n\nChange-Id: Ie8e2b03f93ace0283c7a759d2404022208c0c553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11a476bbc4641ea1d6972ee15bd0813727028807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11a476bbc4641ea1d6972ee15bd0813727028807"}]},"branch":"refs/heads/master"},"6a1f7f299b2099814399938e5c23b0397f2caa15":{"kind":"REWORK","_number":2,"created":"2021-02-26 06:22:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/17/777717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/777717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/777717/2"}}},"commit":{"parents":[{"commit":"b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7","subject":"Add a config file option to swift-manage-shard-ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:22:41.000000000","tz":660},"subject":"Add name to verbose compact output","message":"Add name to verbose compact output\n\nWe have a verbose option in s-m-s-r and when you get to the final\ncollapse into root with the compact command, it can look a little\nscrewy:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the\n  sharder runs.\n\nSo if you pass the --verbose flag the name is added to the output:\n\n  $ swift-manage-shard-ranges --verbose  --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027AUTH_admin/shardme\u0027\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nThis also works any time you use the compact command:\n\n  $ swift-manage-shard-ranges --verbose  --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-0\u0027\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027e9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-1\u0027\n    objects: 0         lower: \u0027e9\u0027\n      state: active    upper: \u0027j9\u0027\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-2\u0027\n    objects: 0         lower: \u0027j9\u0027\n      state: active    upper: \u0027o9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-3\u0027\n    objects: 0         lower: \u0027o9\u0027\n      state: active    upper: \u0027t9\u0027\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-4\u0027\n    objects: 0         lower: \u0027t9\u0027\n      state: active    upper: \u0027y9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n    objects: 0         lower: \u0027y9\u0027\n      state: active    upper: \u0027\u0027\n\nChange-Id: Ie8e2b03f93ace0283c7a759d2404022208c0c553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a1f7f299b2099814399938e5c23b0397f2caa15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a1f7f299b2099814399938e5c23b0397f2caa15"}]},"branch":"refs/heads/master"},"4818b5731def22f52feb869c6249c2d7427859d5":{"kind":"REWORK","_number":3,"created":"2021-02-26 06:25:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/17/777717/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/777717/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/777717/3"}}},"commit":{"parents":[{"commit":"b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7","subject":"Add a config file option to swift-manage-shard-ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:25:35.000000000","tz":660},"subject":"Add name to verbose compact output","message":"Add name to verbose compact output\n\nWe have a verbose option in s-m-s-r and when you get to the final\ncollapse into root with the compact command, it can look a little\nscrewy:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the\n  sharder runs.\n\nSo if you pass the --verbose flag the name is added to the output:\n\n  $ swift-manage-shard-ranges --verbose  --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027AUTH_admin/shardme\u0027\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nThis also works any time you use the compact command:\n\n  $ swift-manage-shard-ranges --verbose  --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-0\u0027\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027e9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-1\u0027\n    objects: 0         lower: \u0027e9\u0027\n      state: active    upper: \u0027j9\u0027\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-2\u0027\n    objects: 0         lower: \u0027j9\u0027\n      state: active    upper: \u0027o9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-3\u0027\n    objects: 0         lower: \u0027o9\u0027\n      state: active    upper: \u0027t9\u0027\n  Donor shard range(s) with total of 0 objects:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-4\u0027\n    objects: 0         lower: \u0027t9\u0027\n      state: active    upper: \u0027y9\u0027\n  can be compacted into acceptor shard range:\n       name: \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n    objects: 0         lower: \u0027y9\u0027\n      state: active    upper: \u0027\u0027\n\nChange-Id: Ie8e2b03f93ace0283c7a759d2404022208c0c553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4818b5731def22f52feb869c6249c2d7427859d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4818b5731def22f52feb869c6249c2d7427859d5"}]},"branch":"refs/heads/master"},"b5b7e27220002c072650c9f8b0367b183c8e6a7a":{"kind":"REWORK","_number":4,"created":"2021-03-01 02:50:16.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/17/777717/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/777717/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/777717/4"}}},"commit":{"parents":[{"commit":"b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7","subject":"Add a config file option to swift-manage-shard-ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-01 02:42:13.000000000","tz":660},"subject":"Add name to compact output","message":"Add name to compact output\n\nCurrently when you get to the final collapse into root with the compact\ncommand, it can look a little screwy:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the\n  sharder runs.\n\nBut it\u0027s fixed if we add the name to the compact output. It also makes\ngives it just a little bit more useful information:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    \u0027AUTH_admin/shardme\u0027\n      objects: 0         lower: \u0027\u0027\n      state: sharded     upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nThis is what a non root collapse would look like with more ranges:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-0\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027e9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-1\u0027\n      objects: 0         lower: \u0027e9\u0027\n      state: active      upper: \u0027j9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-2\u0027\n      objects: 0         lower: \u0027j9\u0027\n      state: active      upper: \u0027o9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-3\u0027\n      objects: 0         lower: \u0027o9\u0027\n      state: active      upper: \u0027t9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-4\u0027\n      objects: 0         lower: \u0027t9\u0027\n      state: active      upper: \u0027y9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-5\u0027\n      objects: 0         lower: \u0027y9\u0027\n      state: active      upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nChange-Id: Ie8e2b03f93ace0283c7a759d2404022208c0c553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5b7e27220002c072650c9f8b0367b183c8e6a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5b7e27220002c072650c9f8b0367b183c8e6a7a"}]},"branch":"refs/heads/master"},"eb321c54c1efcd9d7fce0e6f7b3e490ce0931576":{"kind":"REWORK","_number":5,"created":"2021-03-01 09:34:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/777717/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/777717/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/777717/5"}}},"commit":{"parents":[{"commit":"b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7","subject":"Add a config file option to swift-manage-shard-ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b019a7aedad0a9fa1c7eec1e9ec70c6145266ca7"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-01 09:34:25.000000000","tz":0},"subject":"Add name to compact output","message":"Add name to compact output\n\nCurrently when you get to the final collapse into root with the compact\ncommand, it can look a little screwy:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the\n  sharder runs.\n\nBut it\u0027s fixed if we add the name to the compact output. It also makes\ngives it just a little bit more useful information:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    \u0027AUTH_admin/shardme\u0027\n      objects: 0         lower: \u0027\u0027\n      state: sharded     upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nThis is what a non root collapse would look like with more ranges:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-0\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027e9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-1\u0027\n      objects: 0         lower: \u0027e9\u0027\n      state: active      upper: \u0027j9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-2\u0027\n      objects: 0         lower: \u0027j9\u0027\n      state: active      upper: \u0027o9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-3\u0027\n      objects: 0         lower: \u0027o9\u0027\n      state: active      upper: \u0027t9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-4\u0027\n      objects: 0         lower: \u0027t9\u0027\n      state: active      upper: \u0027y9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-5\u0027\n      objects: 0         lower: \u0027y9\u0027\n      state: active      upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nChange-Id: Ie8e2b03f93ace0283c7a759d2404022208c0c553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb321c54c1efcd9d7fce0e6f7b3e490ce0931576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb321c54c1efcd9d7fce0e6f7b3e490ce0931576"}]},"branch":"refs/heads/master"},"fd5bd9bfbd08a57d8f87ed331526fa0907e98616":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-03-05 17:19:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/777717/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/777717/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/777717/6"}}},"commit":{"parents":[{"commit":"1ec19030a7c44620e9615c7e41efe4552421d341","subject":"Add a config file option to swift-manage-shard-ranges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ec19030a7c44620e9615c7e41efe4552421d341"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-05 17:18:50.000000000","tz":0},"subject":"Add name to compact output","message":"Add name to compact output\n\nCurrently when you get to the final collapse into root with the compact\ncommand, it can look a little screwy:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the\n  sharder runs.\n\nBut it\u0027s fixed if we add the name to the compact output. It also makes\ngives it just a little bit more useful information:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    \u0027AUTH_admin/shardme\u0027\n      objects: 0         lower: \u0027\u0027\n      state: sharded     upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nThis is what a non root collapse would look like with more ranges:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-0\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027e9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-1\u0027\n      objects: 0         lower: \u0027e9\u0027\n      state: active      upper: \u0027j9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-2\u0027\n      objects: 0         lower: \u0027j9\u0027\n      state: active      upper: \u0027o9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-3\u0027\n      objects: 0         lower: \u0027o9\u0027\n      state: active      upper: \u0027t9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-4\u0027\n      objects: 0         lower: \u0027t9\u0027\n      state: active      upper: \u0027y9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-5\u0027\n      objects: 0         lower: \u0027y9\u0027\n      state: active      upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nChange-Id: Ie8e2b03f93ace0283c7a759d2404022208c0c553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd5bd9bfbd08a57d8f87ed331526fa0907e98616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd5bd9bfbd08a57d8f87ed331526fa0907e98616"}]},"branch":"refs/heads/master"},"a045f90b4d7ccc2da4cd52883c04c176b9ec920f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-12 19:16:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/17/777717/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/777717/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/777717/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/777717/7"}}},"commit":{"parents":[{"commit":"66bd11bde104012a508bdc510bc297df8810e0d5","subject":"Merge \"Add a config file option to swift-manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66bd11bde104012a508bdc510bc297df8810e0d5"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-26 06:14:30.000000000","tz":660},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-12 19:16:30.000000000","tz":-480},"subject":"Add name to compact output","message":"Add name to compact output\n\nCurrently when you get to the final collapse into root with the compact\ncommand, it can look a little screwy:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    objects: 0         lower: \u0027\u0027\n      state: active    upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    objects: 0         lower: \u0027\u0027\n      state: sharded   upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the\n  sharder runs.\n\nBut it\u0027s fixed if we add the name to the compact output. It also makes\ngives it just a little bit more useful information:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614319760.63229-5\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027\u0027\n  can be compacted into acceptor shard range:\n    \u0027AUTH_admin/shardme\u0027\n      objects: 0         lower: \u0027\u0027\n      state: sharded     upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nThis is what a non root collapse would look like with more ranges:\n\n  $ swift-manage-shard-ranges --conf /etc/swift/container-server/1.conf /mnt/sdb1/1/node/sdb1/containers/925/c2a/e77a449c51d01ef47471efb208c20c2a/e77a449c51d01ef47471efb208c20c2a.db  compact\n  Loaded db broker for AUTH_admin/shardme.\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-0\u0027\n      objects: 0         lower: \u0027\u0027\n      state: active      upper: \u0027e9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-1\u0027\n      objects: 0         lower: \u0027e9\u0027\n      state: active      upper: \u0027j9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-2\u0027\n      objects: 0         lower: \u0027j9\u0027\n      state: active      upper: \u0027o9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-3\u0027\n      objects: 0         lower: \u0027o9\u0027\n      state: active      upper: \u0027t9\u0027\n  Donor shard range(s) with total of 0 objects:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-4\u0027\n      objects: 0         lower: \u0027t9\u0027\n      state: active      upper: \u0027y9\u0027\n  can be compacted into acceptor shard range:\n    \u0027.shards_AUTH_admin/shardme-7820d6e9550a1661e01e0538aea8cc1b-1614565810.95419-5\u0027\n      objects: 0         lower: \u0027y9\u0027\n      state: active      upper: \u0027\u0027\n  Once applied to the broker these changes will result in shard range compaction the next time the sharder runs.\n  Do you want to apply these changes? [y/N]\n\nChange-Id: Ie8e2b03f93ace0283c7a759d2404022208c0c553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a045f90b4d7ccc2da4cd52883c04c176b9ec920f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a045f90b4d7ccc2da4cd52883c04c176b9ec920f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
