)]}'
{"id":"openstack%2Fswift~777890","triplet_id":"openstack%2Fswift~master~I3da9b59f6744b10853d89b51ae15f05bdd51d02d","project":"openstack/swift","branch":"master","topic":"p-sharder-compact-to-root","hashtags":[],"change_id":"I3da9b59f6744b10853d89b51ae15f05bdd51d02d","subject":"sharder: only shrink to root if all ranges are involved","status":"MERGED","created":"2021-02-28 16:01:03.000000000","updated":"2021-03-15 19:31:57.000000000","submitted":"2021-03-15 19:30:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"777890-p-sharder-compact-to-root","meta_rev_id":"ec6f1562acec3c795cf98dc708fd18b0028f533c","_number":777890,"virtual_id_number":777890,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 19:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-03-01 02:17:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2021-03-15 15:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2021-03-15 15:19:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-28 17:14:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 02:17:34.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-03-15 15:19:35.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"60ce7b01289fdd5b23177a638842bc85fef53e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-02-28 16:01:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"719ab1d3b6bcb8e1cf7881c28628062baed53071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 17:14:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1493b70b16b4b3482b5dad4ed39a1b0 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/8ce6f2c009524f098b352720156c145d : SUCCESS in 1h 12m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bdcaff19e12b46c0902a595c83dc6d75 : SUCCESS in 43m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c53066a507084686ad1a2afbc995c6dd : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1b3052de0d764b549a843eebb7a88ec9 : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad61fac73f9d4ac181c0330dcc4989d2 : SUCCESS in 26m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a09d52da10d44001bdfdc6e2ff45b356 : SUCCESS in 27m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65b655a81c614d238d5bf068419165e4 : SUCCESS in 27m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6404391a942949c297ec1f2981aaf775 : SUCCESS in 28m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1d8beee14c42e5936bfc561952fead : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a81acbf3a444edfb4797663b3f74fea : SUCCESS in 23m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f2b804ac4f24277ad73004ed66451d0 : SUCCESS in 30m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1765efc9cbb04f44acbd179dd4448851 : SUCCESS in 31m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/69d08d58cbc9484d822ba932fc909fb0 : SUCCESS in 30m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a5ed505fb8a0433f8beff3126406fc11 : SUCCESS in 24m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0e935ea5abff4f3d998ac84292e4b762 : SUCCESS in 24m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e91167e5b8f4323ae62ae2d0fc44787 : FAILURE in 28m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/78ef4210b9e643a597e4fdfb71e9b581 : SUCCESS in 51m 39s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b02dcf6b0a7f4d95a5f5ff6b77560f23 : SUCCESS in 43m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/472590c68ddd43b5a9a1b0a8d56766c4 : SUCCESS in 33m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/830479fe1a244b17b51c69316d04f56c : SUCCESS in 41m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2338f642ec141a28a651c4cdfcb435e : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868812e48323476795f0abc2fad4d802 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45afe7f690a84261a110012785e61108 : FAILURE in 11m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fc60fab69f146219d5393c39a5f7303 : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1cf895b62dd7d80742e98e1a72ca8e0a2c77a0ba","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-01 02:17:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7cc4752cd07242883d23acb51dc573021471ec8f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-15 15:19:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"088aa12800773383268baf5870034a1d44163e2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:21:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55deb2a9b033eb06da9349a4af4a51c704e093dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:30:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f917c325fc4e448b842bc07698bd0227 : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/2907206d4815466dbb25ff3fce4e3d8f : SUCCESS in 1h 18m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e0ea6d1fc6f4ef68931772860444764 : SUCCESS in 39m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/45291550a7274ec3938112acd8b61ae5 : SUCCESS in 12m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ceacd6a719894824a91cecbe094d3549 : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/058576d93dab406ca3af9a82f2df5bfe : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eef19c7e39454ed1a3cd928a7f334be4 : SUCCESS in 16m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5ec88db96e1460095abb64dbbc6281d : SUCCESS in 17m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9dfb958312a246acae1a9e84dbc3e981 : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a5aeca0db3343798c3e6bef6d0fa8c8 : SUCCESS in 22m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82d6d058b23d4da688d3274e27e07486 : SUCCESS in 23m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df39c16cc62b4e75a4b60efbe11f9cfb : SUCCESS in 21m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/481049f1ad2a4097a37836ef4f819de8 : SUCCESS in 25m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/33078784cead4339ba7c0d2194391327 : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/30889b563b0a4b8bbf88104a2227dc7c : SUCCESS in 18m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f1c973e060474e07961aeaa160913c03 : SUCCESS in 19m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3522710c7ad046e892b9296237b9f5df : SUCCESS in 6m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a4c414ce04f490c9a0d4cf91a9afc70 : SUCCESS in 47m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a7101d59a5e14e529c425406c4d09022 : SUCCESS in 52m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7199e8d891a3441cb0cfcd2d618b3a5a : SUCCESS in 41m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d14205eb6c443ff92af7f281ccc1602 : SUCCESS in 44m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab01217492444f329258a3af20ce5a20 : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2afac7e2ee8f46bf83b5665f53d52b8f : SUCCESS in 6m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18e90748c58e4cc297f06cf744faf5b3 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0cc9772fd408594ac51629de92a46a9044a07896","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:30:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec6f1562acec3c795cf98dc708fd18b0028f533c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566f5b2bc3fe4ffea96587aca7295e1f : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a48b0875732740b8b0e6d10f0bd200d8 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4b7d05f08a6d48c2b4c8f0c2c409da35 : FAILURE in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01dec1240e58c9f94094ffa9a9675e908d57a049","revisions":{"01dec1240e58c9f94094ffa9a9675e908d57a049":{"kind":"REWORK","_number":1,"created":"2021-02-28 16:01:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/777890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/777890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/777890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/777890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/777890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/777890/1"}}},"commit":{"parents":[{"commit":"41f85f3969d854dceb1a0f6f22c377d62e7e6477","subject":"sharder: fix loop in find_compactible_shard_sequences","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41f85f3969d854dceb1a0f6f22c377d62e7e6477"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-28 15:38:17.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-02-28 15:57:51.000000000","tz":0},"subject":"sharder: only shrink to root if all ranges are involved","message":"sharder: only shrink to root if all ranges are involved\n\nTighten up conditions for shrinking shards to their root container so\nthat *all* shard ranges must be included in the single compactible\nsequence. Previously shard ranges in states that were not eligible for\nshrinking (such as CREATED) would not prevent other shards shrinking\nto root.\n\nChange-Id: I3da9b59f6744b10853d89b51ae15f05bdd51d02d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01dec1240e58c9f94094ffa9a9675e908d57a049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01dec1240e58c9f94094ffa9a9675e908d57a049"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
