)]}'
{"id":"openstack%2Fswift~780114","triplet_id":"openstack%2Fswift~master~I0f537209bce87a0b7a78b33813b694a6b697104a","project":"openstack/swift","branch":"master","topic":"p-relinker-cleanup","hashtags":[],"change_id":"I0f537209bce87a0b7a78b33813b694a6b697104a","subject":"relinker: continue cleaning up old files despite failure","status":"MERGED","created":"2021-03-11 21:21:27.000000000","updated":"2021-03-12 22:40:57.000000000","submitted":"2021-03-12 22:38:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"780114-p-relinker-cleanup","meta_rev_id":"a078583e32526d248a2c6be5050c2b9340d5c54c","_number":780114,"virtual_id_number":780114,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 22:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-12 22:40:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-12 16:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 22:12:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-03-11 22:37:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb891f87d78eefb4a30b2ee00b552c221900020e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-11 21:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cc2be1b4bc583a91dce67ceef3ee8cc6fc2e1e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-11 22:12:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nBetter to remove what we can.","accounts_in_message":[],"_revision_number":1},{"id":"6a262f5488fe2f56d6d55892c0f2d6b36b63f4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 22:37:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3253b4474d544b9e85519e46fd804972 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/cfc6b706217b40c994dc15aa0dd05adf : SUCCESS in 54m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/12493c5fe79945e595bb1f02b78c9d58 : SUCCESS in 46m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8dcacfa42d9438fb8ee494d103e064e : SUCCESS in 11m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/84b6535401ab410981ff275bb33419fe : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81269e7559ec497e9ae8c0ee81a52b66 : SUCCESS in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb6a91086c3b4894b3bf6dad0187735b : SUCCESS in 17m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1d37086bf2fe4262aedc5ad3009537a7 : SUCCESS in 15m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf246904a9c42159b9ea7f6f07657cb : SUCCESS in 23m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/152768514bdb48829027216018977948 : SUCCESS in 16m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/05357171b63f467a81124ed87b1e61d7 : SUCCESS in 30m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03c9410876b049f9b3616e8a42f1676a : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9804ae3d291f4023866752842fbb317f : SUCCESS in 16m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b06495e8b87a49ccbc611fab9763e316 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1b5c94213dfe43d08a192345ae78ebba : SUCCESS in 17m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6613b8d7296840569d1c91224b2c8632 : SUCCESS in 23m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ae873094a204e93818adf08e2a47d49 : SUCCESS in 11m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76b253d61ce14b7cbe68c82cfa9cb166 : SUCCESS in 26m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/252fc74bffd741c19881d69c75df39cf : SUCCESS in 46m 32s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8e1f3b31e47947a38558e31f3a514b4d : SUCCESS in 55m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22bb14104d60418cb95a6cdb8139c4ec : SUCCESS in 44m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea0450476fdf4cdaba1438ba016fad9d : SUCCESS in 42m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bb165492e794fb187e944a6485c834e : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a97ec8bb3d4e6e97ec1297ba0a3a43 : SUCCESS in 6m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0558e481fc624b37a25e2bb733c23e26 : FAILURE in 11m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/778630cb19444e5a803e59373f983e15 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":1},{"id":"618317b53afff8469151c558e65f58562f97068a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 22:37:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dffb12329c5d145aad5362da1234e054978654ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 23:42:09.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e316eccebe8c7fe61728156ea6baa71ec04e8c47","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 01:07:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd73219ca013fc23a0cbdbc0c221f5e0d8797105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 02:04:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f18bfa7453d4ed0bc1915c7a9fc3c57 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/ffce4c59b94944b98a55d4fd4dc5a90c : SUCCESS in 54m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ecb7ad017b4c445fb1a641c1d24f3a8e : SUCCESS in 50m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8044cac4b27b4caf9a995595e2fd62ce : SUCCESS in 11m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/948073fc479d44d893ed6cd7b01863ce : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/779a93f1fa4742228b0e8f1abfad136f : SUCCESS in 13m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1d42b74ad864f38b5ef5620512bd138 : SUCCESS in 13m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d94b85874e5d476b99947488d06a5e6b : SUCCESS in 18m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/92a5720f0d8246f98703d06ad6ed0dfb : SUCCESS in 14m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb896d4930084550922fc2e621742183 : SUCCESS in 11m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b2b3d9040e0445efab83545de90f45a3 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6e18f7bca41d42a48094abc4e1c7b37f : SUCCESS in 21m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cece9b8248f74a9e99403ed2550349cd : SUCCESS in 17m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7c559ade6f39472fa173b1f6734d5365 : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cf7697c7375d4985ac000cb9e44176e9 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/79ca3c12cd434878a208ced0fb2f8e3e : SUCCESS in 18m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4bf4b1a404ce4e34b7b244636e4575c2 : FAILURE in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e84be41b57a04e918dde5769b4ed759e : FAILURE in 21m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bebca10724ed4a22af0ac0af7188243a : SUCCESS in 51m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/591d3ed519f7494787e626759bd62c33 : SUCCESS in 52m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0144accfaafc408aa29f41d4072ebb46 : SUCCESS in 41m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b121b41bbaa940af890ba643bb7b0ffd : SUCCESS in 38m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6d22b74853641da98b9be9342f0a3f2 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f88bb30cb44f1aa839788fbd2aad1b : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b60a6c98368487bb14012208b90a933 : FAILURE in 12m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/700d5c8c186e4600bdd2f9b02d53c34b : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":1},{"id":"010ff82e6a34d0499281461f57f724838e6dda4b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 03:23:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70a1d2d374284ef9dce9aecb093f700cd0ba0f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 04:39:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b08afc41796429a822f5fe12e45719c : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/4279c51c3bac4869a6edfe3044a0b597 : SUCCESS in 1h 06m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c57373936f9416b8c91d6349852c9af : SUCCESS in 50m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f13015d3d0e14fa3ab25e54db19917aa : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a67881942fc64f1a8320d0a879c29075 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a2942513be34659a0ce10fd085dcb91 : SUCCESS in 13m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d97970cc9374a3fb4e801d3b18657b9 : SUCCESS in 23m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/58c20c0e2a1d44758a5b50fbf7c74a28 : SUCCESS in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2f8603b1881d450e8b017ac33643eff0 : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3dbe20ffbb224f6089d303edea46dc5b : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/679c9e2e84534461a8a70577b28a01b0 : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44cd73fc982348978842c8df5af9177e : SUCCESS in 20m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a781a2235034fcfae491cd506667fb5 : SUCCESS in 25m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b7ec5c95f80d466a8cc0f2f79a9be5cd : SUCCESS in 20m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7587f24d24d847f8a4e90fcc4bb7754e : SUCCESS in 21m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/af70dc16a4b54631851b21fee712d518 : SUCCESS in 26m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c05eabeea79e41e5b1300f68330dd90b : SUCCESS in 5m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5cb4d617bdc44107906bbf73486791e5 : FAILURE in 23m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6fbbe955dbe246bf9f847251824d7d52 : SUCCESS in 46m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/67e835728edc4ab7ba568d8d6b7e449d : SUCCESS in 46m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee61738cd0f242d1ac5cb57b7d817894 : SUCCESS in 34m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec3284fc12444ee5827bcaa8b9e2f1f4 : SUCCESS in 39m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd6401c73ab64a038882dc44d1a0e487 : SUCCESS in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76d2c8d7a094a81aa5b515f2cb2a44a : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc43ae7d5d0546058021abd082a21136 : FAILURE in 12m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e8dc462698148d1baeddfe35664fea6 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7d9517242b15a12be4cf490e2a0b5faa8d667f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 07:02:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de87901d82b10eafe2448ddae4b416d2860a9018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 08:09:00.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"75e45e87bbbb258256effa5cc7c87059c56d27cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 16:22:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"65edd41aee36f488abe5c07650635151c4b209df","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 16:22:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ccb9597c9923b5b7810ca9c1b229b3e2f7049f6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:48:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121f1bdf965841cdad2e568bd0d72d58 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/078ae773fe2b4c3984cce0bfd36247f7 : SUCCESS in 1h 13m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db8589b4e81f46be950b8ffaece68a6b : SUCCESS in 44m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/916055754d2d43ce9cf0f3c3702d036d : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c565b47906214f19907029c2424f27d4 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1643baad36e84f048024b5b50a193570 : SUCCESS in 14m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73937119add4405e8637c72813056e46 : SUCCESS in 16m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f310065281c44b6b0951d24ccea4b5b : SUCCESS in 18m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/860dfb14012e41709001508cf722981a : SUCCESS in 15m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b968a37cc54c7b8c7bee3e4cc9731d : SUCCESS in 16m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ada00fd779d4822916f1f6802651e05 : FAILURE in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2bcd0088c3684c80b00087febd5c4a17 : SUCCESS in 24m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/355d21f1e5e24a2399aa5f96824ee7bc : SUCCESS in 32m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7911ee4c7f7648f1ab7bf34a89d5e004 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/70dbd1d7dfaa445788a2a966c0376180 : SUCCESS in 34m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5d9941188656443e8575b326c2c40d1e : SUCCESS in 21m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2547e5eb7aa474fbb8a4c83276d489b : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/933ed149f7894a909679ebee61b929a9 : FAILURE in 29m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f7fe99acd2e4e2abfb4cf4562600422 : SUCCESS in 51m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6181419d9ecc43008711232e702c005c : SUCCESS in 49m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b14caec67644d08b8ec923e9e5666d1 : SUCCESS in 46m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24cf5be681df4744bd07caadff8727c7 : SUCCESS in 42m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c56dbf85bfe478a9c301ec1f07ab306 : SUCCESS in 30m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9245e12f6846beb503a689f411e139 : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7db0f5b6266741ab910dcc3cce713609 : FAILURE in 10m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e50dbfd3b0d84dfa870a867417fb816d : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f0915649d7f6204578a7bc22e4ce755221e07980","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-12 19:21:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bcd3bd8b766af94a73aa7d8434793c1f5f68a298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:37:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94909ddf0aa44525a6af743493028efe : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/4eedfc1d674445bf880942415fbf17f4 : SUCCESS in 1h 09m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/00cf2d260de44616a4cf7b803b220d26 : SUCCESS in 51m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05773fbc05504ebbafaabc8da9d65bfa : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f9ea1c642ad5444ca1b61c23600068a8 : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b92e4535a7294bbc957d2893a45e9448 : SUCCESS in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bfcfae385a164b07a80df02fbf8ec6db : SUCCESS in 16m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9b837711561547a5a685521eec8948f9 : SUCCESS in 16m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/918033dd466b4e83ba49b039e3ef7939 : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04fa3e918b7440148522fb12902945dd : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c595fc8e8c84f3894fc176fae050c69 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0876c90acc34921a4c21ec4700af76a : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b76705a3ba1d429cbf90f73eb3e9e721 : SUCCESS in 18m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4fac66149d2040dfa08fd712228c050f : SUCCESS in 19m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/488a01d2cdce42cd984db427a1c0e3c3 : SUCCESS in 20m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/caa37d30d8b34722b661896bbab32747 : SUCCESS in 21m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b82a25ae5ef246c2b0948f7ffa0febaa : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65d0648cebaf402bbb67ce765f4bdc8b : SUCCESS in 21m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e8f269ac2a914f95908a94defb9583e7 : SUCCESS in 51m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/59db738a022343b18596d8b1be693922 : SUCCESS in 45m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a960f21bcbc499e9fcb9a1e96f08623 : SUCCESS in 42m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfb3132226b4439792fade238af7c1c0 : SUCCESS in 41m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c1438be5114fb5a056356fdf365be2 : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39826dcdcfbe4cf7809d8ecf6fa52cfc : SUCCESS in 6m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a34626f0f6fa47b6919cd230d21b4d42 : FAILURE in 13m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec31a4b437ba4149b3ea49a7cf3f523c : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":2},{"id":"364362366508ac8de81427b289c5345e55783ebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 21:11:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b1437f547bd341f8f9fe0b8c7f058aa9254f7c4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:38:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d608a2d26c4d5589ac43d4211078bf : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5014537c418041d6929a1fa0511be998 : SUCCESS in 1h 20m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/951fa9d64f2245ccbaef6781b6c9b0bc : SUCCESS in 53m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ed9137c0ce31423baca69c86d68c7c49 : SUCCESS in 11m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2f628b8f11f84edc8abbdb2ffbc2ee34 : SUCCESS in 10m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a53e59a369149fcbf9a2ed8b9332bbd : SUCCESS in 21m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dddeb5f7631c42e294135a7d4f949bd5 : SUCCESS in 16m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e2923750827472d8e248112f1da2940 : SUCCESS in 21m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8117d8c325fc4a5e96895a6a8f7e8530 : SUCCESS in 14m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aab789e612d049ac8966812f229313a2 : SUCCESS in 15m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba57db2440574ac2b6085b13050bf075 : SUCCESS in 24m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4b72557d0e44b57a13577274eeab608 : SUCCESS in 21m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90be2d320f1d48f8896bbcbdba2ce50b : SUCCESS in 17m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/14823dbb41fb41349b12f37d232b6a93 : SUCCESS in 20m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9801ff5e42d1409dba48460916a43269 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/baabe5721deb4a3aae5301d6a45880e7 : SUCCESS in 23m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45bcf37a94e14df6b98e526010346ca8 : SUCCESS in 9m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fc61b7cf4824485f9021f167c0a7d340 : SUCCESS in 46m 03s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2d8630a80e054a88b5b76f17c4eaf6a2 : SUCCESS in 51m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca4e5c956c7f4725a4f20b86e77e187c : SUCCESS in 38m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56d15ee961d246338620e3413bbdf37f : SUCCESS in 44m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9a414ad5d544bddb456d827843e5a9e : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b67e96d12d54e30865570ae34d91e15 : SUCCESS in 7m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82321cea33a74380add451c51e0b4320 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ea89d144e9c13a288732e019743a495bb76ab239","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:38:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2cf62a3b60e72ae50fc19efc99cd4ace54861fed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 22:40:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326aaa3413de4866bd12ff15eca3f524 : SUCCESS in 2m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/88b95575aa3442f3b7f6d40d22a706c0 : SUCCESS in 1m 13s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0a2c04d3cc7a4341aae0ee1fe1af6c0f : SUCCESS in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82f3d0ff97b047e2f5f736489dd4ff8023576d3c","revisions":{"9e2609533edb39affbe5f288e5df42d3acf0b42a":{"kind":"REWORK","_number":1,"created":"2021-03-11 21:21:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/14/780114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/780114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/780114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/780114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/780114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/780114/1"}}},"commit":{"parents":[{"commit":"228d44209790775f4b0ad4041a4d01440da44653","subject":"relinker: make cleanup checks more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/228d44209790775f4b0ad4041a4d01440da44653"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-11 21:03:01.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-11 21:12:27.000000000","tz":0},"subject":"relinker: continue cleaning up old files despite failure","message":"relinker: continue cleaning up old files despite failure\n\nIf the relinker cleanup fails to remove one file in an old partition\nhash dir then it will now continue to attempt to remove other files in\nthe same hash dir.\n\nAlso, adds and improves unit tests for relinker cleanup.\n\nChange-Id: I0f537209bce87a0b7a78b33813b694a6b697104a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e2609533edb39affbe5f288e5df42d3acf0b42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e2609533edb39affbe5f288e5df42d3acf0b42a"}]},"branch":"refs/heads/master"},"82f3d0ff97b047e2f5f736489dd4ff8023576d3c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-12 16:22:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/780114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/780114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/780114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/780114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/780114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/780114/2"}}},"commit":{"parents":[{"commit":"ebee4d45552312a688909ef109588a2e69ab29cc","subject":"relinker: make cleanup checks more robust","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebee4d45552312a688909ef109588a2e69ab29cc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-11 21:03:01.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-03-12 16:22:20.000000000","tz":0},"subject":"relinker: continue cleaning up old files despite failure","message":"relinker: continue cleaning up old files despite failure\n\nIf the relinker cleanup fails to remove one file in an old partition\nhash dir then it will now continue to attempt to remove other files in\nthe same hash dir.\n\nAlso, adds and improves unit tests for relinker cleanup.\n\nChange-Id: I0f537209bce87a0b7a78b33813b694a6b697104a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82f3d0ff97b047e2f5f736489dd4ff8023576d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82f3d0ff97b047e2f5f736489dd4ff8023576d3c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
