)]}'
{"id":"openstack%2Fswift~781084","triplet_id":"openstack%2Fswift~master~I2a549ce829eb0bc38406575582202e1d8dd1a0e2","project":"openstack/swift","branch":"master","topic":"p-cors-test-headless","hashtags":[],"change_id":"I2a549ce829eb0bc38406575582202e1d8dd1a0e2","subject":"Use headless option for cors functional test runner","status":"MERGED","created":"2021-03-17 11:50:28.000000000","updated":"2021-03-18 06:53:01.000000000","submitted":"2021-03-18 06:48:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"781084-p-cors-test-headless","meta_rev_id":"e13b3f965861ce2965a6fe64d4964ae373b332b1","_number":781084,"virtual_id_number":781084,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 06:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-17 23:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-17 23:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 13:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 17:48:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-03-17 22:09:04.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"8a25d2863484545f2e0e3639b8ca2349c6b070b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-17 11:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0242fd85225ea25c053f2012a8156f23ff0b8997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 13:02:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732b3b172a0843f8b90c2f03d5255fc8 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/16db5115473246008d9689e4a5235d51 : SUCCESS in 1h 06m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f9e9b0d75eb4607b2c48530ba4f2996 : SUCCESS in 52m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0a0c9e12e18468998de2df8d403cfe5 : SUCCESS in 10m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8d2bccef38564d0d96b16f49d53a9823 : SUCCESS in 7m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e08dc0ffd6f4b7abf343ecacc0e688a : SUCCESS in 15m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/facea2020e95442080442c257224c1cf : SUCCESS in 17m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6be7a7cacffb4e1cb0330f85f6e58d70 : SUCCESS in 15m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3dd5c977912c4435bfc3c0badcb058d8 : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/578b6ca086b9428fa20650a9633a0def : SUCCESS in 18m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1966eb4b497433d87033cd905745fe6 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d006cb3dc9244b9caffa55713262a16a : SUCCESS in 19m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c5d0e0470ee4f07906c0091197c3561 : SUCCESS in 18m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/11e29e77e33f4a6b852c97d701dab600 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8b9a0f081bca4dd0a8f3b843857ea26f : SUCCESS in 19m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8a3bee9a457547c29503ac7c95c2a306 : SUCCESS in 19m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29d0b1dfe96e4bc4941c2df1e5ed52da : SUCCESS in 8m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b35c7f1a7bee4d1aa246d65cd2354ed2 : FAILURE in 26m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1517bc4094bd4d35b8bb2e1f5c8e52d2 : SUCCESS in 51m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/37c21831a5c84ef2a662d849461e4c79 : SUCCESS in 50m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a11f69384db44db9fd3e4441e8ca8ed : SUCCESS in 41m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14af3a108ab54a5a88c0e19087a82664 : SUCCESS in 44m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d90a3593aa5406eb8a805bb55b318c5 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b592a67d947f4ce1b908a4b62f36cd56 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4452d11876db438095dd76d7306be0a9 : FAILURE in 10m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/361fe05993c24d6fb6c3075f99af39f4 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":1},{"id":"feb0d2eb54fad97830f3d7478ad77a0fe1bb0067","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-17 17:48:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29a4eb3fba5c4d072026793856a88a0c9d80160c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-17 17:52:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19c8c5c04de6c96819ee4a55dab1cfc432f43092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-17 22:09:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48a65e0c19219c1fa0628a54b5bc4920e9a1081e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-17 22:12:17.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"31d68587b1f2b8407370391b1ba7c796b5d65dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-17 22:57:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0458d8b1ad93f689ec70342940c51a2d142d8eeb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-17 23:58:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNot sure if it\u0027ll fix it, but seems worth a shot!","accounts_in_message":[],"_revision_number":4},{"id":"7704e8c46e7c0dc8507786e3a4f55b5ff32bab1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:08:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e38a9f34fa4bfab2e90779b187a623 : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/e73ba175f317427399098095ab67e1c5 : SUCCESS in 49m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8cd24ecf773042c1a834ce34b9ed190d : SUCCESS in 39m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db2a15387fe446e482a1600083967b37 : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ab6825ec3097443aabba9aed4ed42abe : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f88c9e968961445ca07ab457838b720f : SUCCESS in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6819293a45f547ebabe1a1525abd83c2 : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af38b096a975498497c45719705dbefd : SUCCESS in 18m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/33abfaf84f7e4ef496be8da33be4e171 : SUCCESS in 13m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/927960344fd54358b7b0218c7e69765d : SUCCESS in 13m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7713093565634934b5f46e2e627dc188 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b9882261973a413987d0246185d9f26a : SUCCESS in 15m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a2b07bf15ea4db58e1de8016532bc32 : SUCCESS in 15m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ddf0e2c42dab4f51b2b743be7531b8b1 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6f11d727ab734190a76bd0e3e4d3610e : SUCCESS in 16m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/791d7031e4964d6386151b1f495d343f : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6761389452b6451eb373b6b65123984c : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b8a52bb7bfb14552aae42d946bd01acc : FAILURE in 28m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b796e01682b3410e874f63b32ddb9189 : SUCCESS in 51m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8f1738350e91494a8cc0e86783346e5a : SUCCESS in 48m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90d0896b952546a799cc21efff2e697b : SUCCESS in 40m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/655eb2a5f6e04441be1f01a3f27674fd : SUCCESS in 29m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/361d80b7272d4736bf246edfe896e631 : SUCCESS in 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640267b579084daf99562f75fc3192e0 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/145d210a69b74bd9a058aaa0ceb92834 : FAILURE in 9m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/886e8e2699e34e018614311dbaf6556b : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":4},{"id":"82cdffb2997de67c19dbd50e222e66e52baca2f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:11:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b320281dbe1543383987dc9f9d8528082aa6f6c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:38:09.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f3962e73585b2717fd20889b268fb096d69df338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 06:48:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90089fe2790841cbb65f5af2bf89a9b0 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/568b2b8eb7d042cfa407fbdc5667e93b : SUCCESS in 1h 05m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1343cbec34714ac9afc172c4223aadac : SUCCESS in 39m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0aca8f07c3ae48d995a3e66f6f39e45d : SUCCESS in 10m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/87248836ea2545aeb149c9a1cd9e62c5 : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be24de654d91492fb47280e2ead6ef95 : SUCCESS in 15m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5246570059e42c6937c74c23d659a3e : SUCCESS in 22m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/181846c2007843a5b61f8363fc67b12c : SUCCESS in 22m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3a47918465094047bfadfa6e44b2e5c6 : SUCCESS in 24m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc00f55479b24d97a721b6f06501e39b : SUCCESS in 19m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d0348f5caaac43a2be32d3804a262beb : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22e05af7c5bf47f6970e339ddcc121ef : SUCCESS in 25m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c658ffeeaa754683b23f6bba96930efb : SUCCESS in 17m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bcf7058737064eac84a15708373d37fc : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/77b538ac10134219905ef898fdccfa26 : SUCCESS in 20m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c2ecbc55d3c5487698e8e55b16c5859e : SUCCESS in 21m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dec8c2a2c3864255b241aef629ca32ef : SUCCESS in 6m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f09252f24cff49c5a3c072eaac20c935 : SUCCESS in 50m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6814ba4fe5814af799d00a488a4eb308 : SUCCESS in 50m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d216078e69b4bd1a798a7cc8c3327dc : SUCCESS in 32m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0fd6ba1a37914a938b1f84b9aabe737b : SUCCESS in 33m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d15bad5406f74969a18d5355e57c7c24 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3041f6e8acb48ab8134e5c3fc28afe1 : SUCCESS in 5m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6ade2907c54469994ec31e517f1a26e : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":4},{"id":"01cc49df61621d8e205d1e517d31effd8deb5281","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 06:48:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e13b3f965861ce2965a6fe64d4964ae373b332b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 06:53:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f46fc49445e45a6abd4ac79921719b0 : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2b59f030aea645a6a5a78975e16264f3 : SUCCESS in 1m 08s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/985a547c5dbf4286a7a2b5dbe5e2ef08 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e572938af349a51ff755d493203fce1d2ef777d5","revisions":{"586a734d0b4dfc72be695dd8b1f0f04b2d80bcf5":{"kind":"REWORK","_number":1,"created":"2021-03-17 11:50:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/781084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/781084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/781084/1"}}},"commit":{"parents":[{"commit":"354a86a80ee6de462c240c5684bf982c806bf7a7","subject":"Merge \"py36: Fix syslog fallback to UDP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/354a86a80ee6de462c240c5684bf982c806bf7a7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-17 11:47:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-17 11:47:24.000000000","tz":0},"subject":"Add --headless option for cors functional test runner","message":"Add --headless option for cors functional test runner\n\nSome selenium webdrivers (e.g. Chrome, Firefox) support a headless\noption so we can expose that as an option for the test runner too.\n\nUse --headless for the zuul job in an attempt to fix\nhttps://bugs.launchpad.net/swift/+bug/1918864\n\nChange-Id: I2a549ce829eb0bc38406575582202e1d8dd1a0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/586a734d0b4dfc72be695dd8b1f0f04b2d80bcf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/586a734d0b4dfc72be695dd8b1f0f04b2d80bcf5"}]},"branch":"refs/heads/master"},"8031c606bf9047eb09898776a815ae4db6627f7d":{"kind":"REWORK","_number":2,"created":"2021-03-17 22:09:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/781084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/781084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/781084/2"}}},"commit":{"parents":[{"commit":"354a86a80ee6de462c240c5684bf982c806bf7a7","subject":"Merge \"py36: Fix syslog fallback to UDP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/354a86a80ee6de462c240c5684bf982c806bf7a7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-17 11:47:24.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-17 22:08:47.000000000","tz":-420},"subject":"Add --headless option for cors functional test runner","message":"Add --headless option for cors functional test runner\n\nSome selenium webdrivers (e.g. Chrome, Firefox) support a headless\noption so we can expose that as an option for the test runner too.\n\nUse --headless for the zuul job in an attempt to fix\nhttps://bugs.launchpad.net/swift/+bug/1918864\n\nChange-Id: I2a549ce829eb0bc38406575582202e1d8dd1a0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8031c606bf9047eb09898776a815ae4db6627f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8031c606bf9047eb09898776a815ae4db6627f7d"}]},"branch":"refs/heads/master"},"461e8981378e156496359d8bd830b76c1d81be44":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-17 22:12:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/781084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/781084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/781084/3"}}},"commit":{"parents":[{"commit":"354a86a80ee6de462c240c5684bf982c806bf7a7","subject":"Merge \"py36: Fix syslog fallback to UDP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/354a86a80ee6de462c240c5684bf982c806bf7a7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-17 11:47:24.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-17 22:10:28.000000000","tz":-420},"subject":"Use headless option for cors functional test runner","message":"Use headless option for cors functional test runner\n\nSome selenium webdrivers (e.g. Chrome, Firefox) support a headless\noption so we can expose that as an option for the test runner too.\n\nUse this in an attempt to fix \"Error: cannot open display: :99\" errors\nseen in the gate.\n\nRelated-Bug: #1918864\nChange-Id: I2a549ce829eb0bc38406575582202e1d8dd1a0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/461e8981378e156496359d8bd830b76c1d81be44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/461e8981378e156496359d8bd830b76c1d81be44"}]},"branch":"refs/heads/master"},"e572938af349a51ff755d493203fce1d2ef777d5":{"kind":"REWORK","_number":4,"created":"2021-03-17 22:57:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/781084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/781084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/781084/4"}}},"commit":{"parents":[{"commit":"354a86a80ee6de462c240c5684bf982c806bf7a7","subject":"Merge \"py36: Fix syslog fallback to UDP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/354a86a80ee6de462c240c5684bf982c806bf7a7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-17 11:47:24.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-17 22:57:02.000000000","tz":-420},"subject":"Use headless option for cors functional test runner","message":"Use headless option for cors functional test runner\n\nSome selenium webdrivers (e.g. Chrome, Firefox) support a headless\noption so we can expose that as an option for the test runner too.\n\nUse this in an attempt to fix \"Error: cannot open display: :99\" errors\nseen in the gate.\n\nRelated-Bug: #1918864\nChange-Id: I2a549ce829eb0bc38406575582202e1d8dd1a0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e572938af349a51ff755d493203fce1d2ef777d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e572938af349a51ff755d493203fce1d2ef777d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
